)]}'
{"id":"openstack%2Fsearchlight~657363","triplet_id":"openstack%2Fsearchlight~master~I881dd037494ffe5e82a7e086337775933641dd4e","project":"openstack/searchlight","branch":"master","topic":"slash","hashtags":[],"change_id":"I881dd037494ffe5e82a7e086337775933641dd4e","subject":"[Trivial fix]Remove unnecessary slash","status":"MERGED","created":"2019-05-06 09:41:21.000000000","updated":"2019-05-07 02:15:35.000000000","submitted":"2019-05-07 02:15:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657363-1557195335944-5e7ed9cd","meta_rev_id":"b81f2f70dd6366d9c2356321fb2137b2908e88ab","_number":657363,"virtual_id_number":657363,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-05-07 02:15:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2019-05-07 01:55:59.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2019-05-07 01:55:59.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 01:55:59.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-05-07 02:15:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5a9e3acf43c8b9495f546d25181df15dfe1c028","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-05-06 09:41:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f34b839e731ded38b46db24cd8ef753a50e8e37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 09:47:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/63/657363/1/check/openstack-tox-pep8/94fc142/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/63/657363/1/check/openstack-tox-py27/6a81b59/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/63/657363/1/check/openstack-tox-py36/4cf4ecb/ : SUCCESS in 5m 08s\n- openstack-tox-py37 http://logs.openstack.org/63/657363/1/check/openstack-tox-py37/9ca2240/ : SUCCESS in 5m 01s\n- openstack-tox-cover http://logs.openstack.org/63/657363/1/check/openstack-tox-cover/4e74da0/cover/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/63/657363/1/check/openstack-tox-docs/c031d4d/html/ : SUCCESS in 3m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/657363/1/check/openstack-tox-lower-constraints/71a294c/ : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a50330f6fff8b7dbf89f3c506c752a985b6a0880","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-07 01:55:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b189019668450f65ce545c549ac52902d543de7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 01:56:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcce6df6a44c23673ae9f163a9dd8d2a13265d97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 02:15:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2d2452f919b71823e9accfd55b4bd018709adf04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 02:15:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/63/657363/1/gate/openstack-tox-pep8/13140ca/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/63/657363/1/gate/openstack-tox-py27/123a475/ : SUCCESS in 5m 53s\n- openstack-tox-py36 http://logs.openstack.org/63/657363/1/gate/openstack-tox-py36/ac9ab4a/ : SUCCESS in 6m 15s\n- openstack-tox-py37 http://logs.openstack.org/63/657363/1/gate/openstack-tox-py37/fdf946d/ : SUCCESS in 6m 22s\n- openstack-tox-docs http://logs.openstack.org/63/657363/1/gate/openstack-tox-docs/6ea283f/html/ : SUCCESS in 4m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/657363/1/gate/openstack-tox-lower-constraints/0e44007/ : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de09b7f40d7174c74da8374f7454e725cb524174","revisions":{"de09b7f40d7174c74da8374f7454e725cb524174":{"kind":"REWORK","_number":1,"created":"2019-05-06 09:41:21.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/63/657363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/63/657363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/63/657363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/63/657363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/63/657363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/63/657363/1"}}},"commit":{"parents":[{"commit":"3374cba06c651f4be196ad2fe8f6435fd20dfd35","subject":"Migrate to opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/3374cba06c651f4be196ad2fe8f6435fd20dfd35"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-05-06 09:39:17.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-05-06 09:39:17.000000000","tz":480},"subject":"[Trivial fix]Remove unnecessary slash","message":"[Trivial fix]Remove unnecessary slash\n\nThis is to remove the unncessary slash when the line doesn\u0027t\nexceed the length of 79, to make the code more readable.\n\nChange-Id: I881dd037494ffe5e82a7e086337775933641dd4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/de09b7f40d7174c74da8374f7454e725cb524174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/de09b7f40d7174c74da8374f7454e725cb524174"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
