)]}'
{"id":"openstack%2Fsearchlight~659501","triplet_id":"openstack%2Fsearchlight~master~I31c8d3b244b93fd7d1731bc79dab6727aa2581ea","project":"openstack/searchlight","branch":"master","hashtags":[],"change_id":"I31c8d3b244b93fd7d1731bc79dab6727aa2581ea","subject":"Cap sphinx for py2 to match global requirements","status":"MERGED","created":"2019-05-16 09:08:01.000000000","updated":"2019-05-31 06:29:19.000000000","submitted":"2019-05-31 06:29:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659501-1559284159223-f81bca3b","meta_rev_id":"aef1aba1d91a5284d57ddfe789272c8c829d79a7","_number":659501,"virtual_id_number":659501,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-05-31 06:29:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":2,"date":"2019-05-31 06:10:07.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"value":1,"date":"2019-05-31 06:10:07.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 06:10:07.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-05-31 06:29:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c3d90e448369292771d303b3e67ee29aa53f009","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-05-16 09:08:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f87afc03b52a53bc609c3c82f7ebbc3e0c9fe129","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-16 09:08:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"d415ef200e8b5c640c936fd2b4691001352dcaac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:14:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/01/659501/1/check/build-openstack-api-ref/1d0fde5/html/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/01/659501/1/check/openstack-tox-pep8/6838508/ : SUCCESS in 4m 33s\n- openstack-tox-docs http://logs.openstack.org/01/659501/1/check/openstack-tox-docs/e4d899a/html/ : SUCCESS in 3m 51s\n- requirements-check http://logs.openstack.org/01/659501/1/check/requirements-check/25421a1/ : SUCCESS in 2m 26s\n- build-openstack-releasenotes http://logs.openstack.org/01/659501/1/check/build-openstack-releasenotes/d4ee93f/html/ : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ec7df4a5e4b630a25c2ef4a058fc5921e04122ef","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-05-16 09:15:33.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks! Trinh.","accounts_in_message":[],"_revision_number":1},{"id":"05483f16f7038708b68200435bd9e82df3001311","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-05-16 09:17:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"022712096cc4551e951d8da0bfa9ab36f696ae5d","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-16 09:18:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":2},{"id":"79bde549468c77c7c5ba31361d1ea23cbe294205","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:47:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/01/659501/2/check/build-openstack-api-ref/184023b/html/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/01/659501/2/check/openstack-tox-pep8/b015fea/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/01/659501/2/check/openstack-tox-docs/8d33c43/html/ : SUCCESS in 5m 26s\n- requirements-check http://logs.openstack.org/01/659501/2/check/requirements-check/7221722/ : FAILURE in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/01/659501/2/check/build-openstack-releasenotes/1dde30b/html/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"87a2b19988b664734a739e64b5caec75b7a48569","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-05-16 11:11:49.000000000","message":"Patch Set 2:\n\nHI Trinh.\nIs the python version 3.4? Reference the openstack/requirements/global-requirements, the sphinx have the low py3 version [1].\n[1]https://github.com/openstack/requirements/blob/master/global-requirements.txt#L457","accounts_in_message":[],"_revision_number":2},{"id":"19205798fbfe53be9c667a314b174c656cc45767","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-17 06:51:47.000000000","message":"Patch Set 2:\n\nHi zhangboye. It should be fixed in the requirement repo. Thanks for pointing that out. I will submit a fix.","accounts_in_message":[],"_revision_number":2},{"id":"d1d7a7edf9efd2f48a87c275d3661d0820d8e55f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 07:01:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/01/659501/2/check/build-openstack-api-ref/e70fd6c/html/ : SUCCESS in 4m 08s\n- openstack-tox-pep8 http://logs.openstack.org/01/659501/2/check/openstack-tox-pep8/e22214a/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/01/659501/2/check/openstack-tox-docs/43ba134/html/ : SUCCESS in 3m 33s\n- requirements-check http://logs.openstack.org/01/659501/2/check/requirements-check/2152c5b/ : FAILURE in 4m 01s\n- build-openstack-releasenotes http://logs.openstack.org/01/659501/2/check/build-openstack-releasenotes/23d8def/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f99876f4e43694a2d1a49af98301d696ecd6284a","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-31 02:14:50.000000000","message":"Removed Workflow+1 by Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"44dd8aec7f0c470d3bc37e71f1ccbd464d7b630f","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-31 02:14:53.000000000","message":"Removed Code-Review+2 by Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"38631240b4c8c412db863772e5955df9c7960854","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-31 06:00:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4a1aa81e49e9bc03e53a15506a307f94e9ca15d","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-31 06:10:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nzhangboye, I discussed with the requirements team, and agree to support py3.4 for now. Eventually this cycle, there will be a tool to mass change to py3.6.","accounts_in_message":[],"_revision_number":3},{"id":"e305004b32f8b36f5a6a024e624aac1d33dc0d88","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-31 06:10:15.000000000","message":"Patch Set 3:\n\nFYI, https://review.opendev.org/#/c/661149/","accounts_in_message":[],"_revision_number":3},{"id":"6bbda3ec1ac0f17b1ff50a05bb29f97bc67b0458","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 06:19:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/01/659501/3/check/build-openstack-api-ref/729f3c2/html/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/01/659501/3/check/openstack-tox-pep8/6b63ada/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/01/659501/3/check/openstack-tox-docs/f3949dc/html/ : SUCCESS in 4m 09s\n- requirements-check http://logs.openstack.org/01/659501/3/check/requirements-check/21c2956/ : SUCCESS in 3m 15s\n- build-openstack-releasenotes http://logs.openstack.org/01/659501/3/check/build-openstack-releasenotes/3f0f257/html/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":3},{"id":"a17f787dfa1fe251dac59a9b2d264ac6f0a56ebd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 06:19:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a1df8a2b5b1f1081b53af209400a4634ff7323ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 06:29:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/01/659501/3/gate/build-openstack-api-ref/028d367/html/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/01/659501/3/gate/openstack-tox-pep8/597fe1c/ : SUCCESS in 3m 59s\n- openstack-tox-docs http://logs.openstack.org/01/659501/3/gate/openstack-tox-docs/17d5b41/html/ : SUCCESS in 3m 29s\n- requirements-check http://logs.openstack.org/01/659501/3/gate/requirements-check/dafba88/ : SUCCESS in 3m 09s\n- build-openstack-releasenotes http://logs.openstack.org/01/659501/3/gate/build-openstack-releasenotes/62411fc/html/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":3},{"id":"aef1aba1d91a5284d57ddfe789272c8c829d79a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 06:29:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4359ac5cec2abcff63e2c6ad2ab86891ce8a8b74","revisions":{"33f47ca695e5982c127b1211be88a9cbf453893c":{"kind":"REWORK","_number":1,"created":"2019-05-16 09:08:01.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/01/659501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/01/659501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/01/659501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/01/659501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/01/659501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/01/659501/1"}}},"commit":{"parents":[{"commit":"8c5073dd0fedcd035b6c29f7dc3f6c81c5535402","subject":"Fix a broken URL link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/8c5073dd0fedcd035b6c29f7dc3f6c81c5535402"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-05-16 08:57:47.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-05-16 09:07:55.000000000","tz":480},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: I31c8d3b244b93fd7d1731bc79dab6727aa2581ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/33f47ca695e5982c127b1211be88a9cbf453893c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/33f47ca695e5982c127b1211be88a9cbf453893c"}]},"branch":"refs/heads/master"},"8f602f806797cbad1a926631eb60581b3fb2f572":{"kind":"REWORK","_number":2,"created":"2019-05-16 09:17:25.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/01/659501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/01/659501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/01/659501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/01/659501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/01/659501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/01/659501/2"}}},"commit":{"parents":[{"commit":"8c5073dd0fedcd035b6c29f7dc3f6c81c5535402","subject":"Fix a broken URL link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/8c5073dd0fedcd035b6c29f7dc3f6c81c5535402"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-05-16 08:57:47.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-05-16 09:17:00.000000000","tz":480},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: I31c8d3b244b93fd7d1731bc79dab6727aa2581ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/8f602f806797cbad1a926631eb60581b3fb2f572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/8f602f806797cbad1a926631eb60581b3fb2f572"}]},"branch":"refs/heads/master"},"4359ac5cec2abcff63e2c6ad2ab86891ce8a8b74":{"kind":"REWORK","_number":3,"created":"2019-05-31 06:00:09.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/01/659501/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/01/659501/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/01/659501/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/01/659501/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/01/659501/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/01/659501/3"}}},"commit":{"parents":[{"commit":"8c5073dd0fedcd035b6c29f7dc3f6c81c5535402","subject":"Fix a broken URL link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/8c5073dd0fedcd035b6c29f7dc3f6c81c5535402"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-05-16 08:57:47.000000000","tz":480},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-05-31 05:59:02.000000000","tz":540},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nWe will still support py3.4 for now.\n\nChange-Id: I31c8d3b244b93fd7d1731bc79dab6727aa2581ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/4359ac5cec2abcff63e2c6ad2ab86891ce8a8b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/4359ac5cec2abcff63e2c6ad2ab86891ce8a8b74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
