)]}'
{"id":"openstack%2Fsearchlight~668732","triplet_id":"openstack%2Fsearchlight~master~I85f3a115fed73df7be3b1d4f3d4ade5c7f6c7292","project":"openstack/searchlight","branch":"master","topic":"f29","hashtags":[],"change_id":"I85f3a115fed73df7be3b1d4f3d4ade5c7f6c7292","subject":"DNM: test patch to use f29 nodeset from devstack","status":"ABANDONED","created":"2019-07-02 18:29:01.000000000","updated":"2019-09-13 18:32:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b50c9bfb7648ec53be74b51c18c416fe6b0bc04e","_number":668732,"virtual_id_number":668732,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-07-02 18:49:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":-1,"date":"2019-07-02 18:29:10.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 18:49:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82e5fb00bb1b82c032a65102e6fd44eda931ae86","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-02 18:29:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a421a05c9618d59b4eef72d4f9540fbdd3b1a92b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-02 18:29:10.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b3f360c8c0f1925aff844dbc0f3a118d8eee196e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 18:49:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/32/668732/1/check/openstack-tox-pep8/7ae18dc/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/32/668732/1/check/openstack-tox-py27/3705f38/ : SUCCESS in 5m 26s\n- openstack-tox-py36 http://logs.openstack.org/32/668732/1/check/openstack-tox-py36/17655d9/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/32/668732/1/check/openstack-tox-py37/fd29a4f/ : SUCCESS in 5m 33s\n- openstack-tox-cover http://logs.openstack.org/32/668732/1/check/openstack-tox-cover/0b1dcf0/cover/ : SUCCESS in 5m 17s\n- openstack-tox-docs http://logs.openstack.org/32/668732/1/check/openstack-tox-docs/370f774/html/ : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/668732/1/check/openstack-tox-lower-constraints/b899530/ : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1832943fcf392947763d91a3696800db05716b39","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-13 18:32:57.000000000","message":"Abandoned\n\nWas just a test patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e758482295a7b3e1b7ba120d85756e28ea9419a","revisions":{"2e758482295a7b3e1b7ba120d85756e28ea9419a":{"kind":"REWORK","_number":1,"created":"2019-07-02 18:29:01.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/32/668732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/32/668732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/32/668732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/32/668732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/32/668732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/32/668732/1"}}},"commit":{"parents":[{"commit":"4359ac5cec2abcff63e2c6ad2ab86891ce8a8b74","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/4359ac5cec2abcff63e2c6ad2ab86891ce8a8b74"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-07-02 18:28:47.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-07-02 18:29:00.000000000","tz":-240},"subject":"DNM: test patch to use f29 nodeset from devstack","message":"DNM: test patch to use f29 nodeset from devstack\n\nDepends-on: https://review.opendev.org/#/c/662529/\nChange-Id: I85f3a115fed73df7be3b1d4f3d4ade5c7f6c7292\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/2e758482295a7b3e1b7ba120d85756e28ea9419a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/2e758482295a7b3e1b7ba120d85756e28ea9419a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
