)]}'
{"id":"openstack%2Fsearchlight~678391","triplet_id":"openstack%2Fsearchlight~master~I8ee0c999cb7563df0f6f102929dc25f36af623cc","project":"openstack/searchlight","branch":"master","topic":"ipv6-only-deployment-and-testing","hashtags":[],"change_id":"I8ee0c999cb7563df0f6f102929dc25f36af623cc","subject":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job","status":"ABANDONED","created":"2019-08-25 06:23:23.000000000","updated":"2020-12-07 18:30:54.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"58776d733ba53f6e500627348c40035f5b7a9074","_number":678391,"virtual_id_number":678391,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},{"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-02 02:19:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},{"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},{"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 01:32:35.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-11-25 12:56:31.000000000","updated_by":{"_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},"reviewer":{"_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},"state":"REVIEWER"},{"updated":"2019-11-25 12:57:00.000000000","updated_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"reviewer":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"state":"REVIEWER"},{"updated":"2019-11-25 12:57:19.000000000","updated_by":{"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},"reviewer":{"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},"state":"REVIEWER"},{"updated":"2019-11-25 12:57:32.000000000","updated_by":{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},"reviewer":{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},"state":"REVIEWER"},{"updated":"2020-04-02 02:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ec148ec1194990cea409fc6fcdb2886d8ce9ccd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-25 06:23:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e4719f6ad205930ee3fb3dfd51f314a42bfef6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 06:23:40.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/searchlight on branch master.  The error was:\n\n  The project template \"searchlight-tempest-ipv6-only\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs\n        - openstack-python3-train-jobs\n        - openstack-cover-jobs\n        - publish-openstack-docs-pti\n        - check-requirements\n        - openstack-lower-constraints-jobs\n        - release-notes-jobs-python3\n        - searchlight-tempest-ipv6-only\n\n  in \"openstack/searchlight/.zuul.yaml@master\", line 17, column 3","accounts_in_message":[],"_revision_number":1},{"id":"d374088aa50731eca2b3980e46fc2790aa6108d4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-25 06:31:08.000000000","message":"Topic set to ipv6-only-deployment-and-testing","accounts_in_message":[],"_revision_number":1},{"id":"1e7825cf09bfe5af006fd753b021cfebc1835578","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-25 06:32:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"683a78fc72733eff692742ae0f0c4bd2f7ee7803","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 07:05:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fb974f626624658adfd58726f0c1e7f : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b69580e46084651b9aabc7c95f9f679 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8568d2146e04ea6a7054b498882eb0a : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3116d02c080d4118b84b293abba8aa9e : SUCCESS in 11m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16fe4ce18aa6433a82cf28fef55329bd : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ac69fe169241609ab2c6d747ccd803 : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f819a760f56e4832ad208d5f040e6148 : SUCCESS in 10m 43s\n- searchlight-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a48fa5b84edc4277ab28c249a2ce84cb : FAILURE in 27m 49s","accounts_in_message":[],"_revision_number":2},{"id":"ee98e6bf035f74c3c40e559556d894fd36aad28a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-25 07:21:35.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"afe0dd24872c9aeb9ccdf558458947b89e8881ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 08:05:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8816e2f92d664f27a4e8cd625714ebea : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8197e1ef9a64a709439a5944b90bfcb : TIMED_OUT in 41m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f02168534f44aa3b5af2d0f9427e3dd : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f73f6749d6bf4a099b8ad2640691e92c : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/feefff9db1f74718b9e93f0fa97466b5 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e126cbc33f84bb0943f3ecf02c0c147 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/316202a8c08b4bdab859236422f1e0db : FAILURE in 8m 22s\n- searchlight-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfe90eac16d34dcd99497d6f80fbd361 : FAILURE in 33m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3c3a7d030d575aaf484900f33c32dd6a87ec9bf0","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-09-17 03:05:49.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0b5c492c2133ad588f71be9a34459542b61098d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 03:39:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da75de6ca9dd45d2b2a495971d35690e : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d9327fa0278497da6c46394425cc78d : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43b20246d2224ec28b3b7662c9f31d62 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34e98e2e69164e62b83bbc2e10871f2a : SUCCESS in 5m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f641f86c109e488d81b6c739b04d73f8 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30fdfe220d1d4488a1ec3fec8cb3a40a : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b474ee23a64d4ea0bb81021cb49bb968 : SUCCESS in 5m 34s\n- searchlight-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac4cb98cf5b447ebb9c1e6ca115cc7aa : FAILURE in 32m 19s","accounts_in_message":[],"_revision_number":3},{"id":"d105e1fbff461734d165fe9ba2643ff9ee1b2db9","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-10-18 01:47:52.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bb7006a314ba4a05e3f21b764280c7d95c162df7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 02:25:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d902acf0533648a2a2f5436fc441516d : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57f07e7c34b44ce692409071f388b9be : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cf3e47c47be4a31ae4b063755e73d5d : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff72cce4403543a387dc835ed45c300d : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68b1128ef4f342cea0420dcfa3edd9d8 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9faa2ef7eecf4a72b2a39d0012a78d18 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92ba2dfb81814a33b1a56b6b1da5928a : SUCCESS in 5m 12s\n- searchlight-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/363e33db62524a66992554c4ea1d6a04 : FAILURE in 31m 56s","accounts_in_message":[],"_revision_number":3},{"id":"715aac491db8c954b17e370c7d0d7fc1d42bef1e","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-10-18 02:48:49.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"26daf46318ba5d005544d04080bffa41ed3c8d02","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-10-18 02:51:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"50726af235bde0e439db7d25b6f16b9efb763bb7","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-10-18 03:01:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e306237be5fb403b378e83c73db972b97c11aef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 03:42:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f257d6bbde47bab0eae2b988f3befb : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f859e6e918774533a3a69732d6704926 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0aa29e7b6304c038d72043211c33803 : SUCCESS in 7m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cca535804861489bb92e4464084fd1c5 : SUCCESS in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68667810953a4258b39c19dbaaae6598 : FAILURE in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4369e80a54594f49881d9f459c78b370 : SUCCESS in 6m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b68b2ced2c3b4d7dad5f43102686b855 : FAILURE in 4m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/465f72119de44b268ae98238bed53792 : SUCCESS in 6m 57s\n- searchlight-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/462ba61c06384f9db2dddf097f8b91c6 : FAILURE in 38m 03s","accounts_in_message":[],"_revision_number":6},{"id":"0d5a9adc4f86f2b4247885f5953bbf3c311ebf46","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-10-18 04:10:38.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"519abbac9f858a45884781b47d79a58ea4939b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 04:45:09.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7874f5bc214fbbae054717629cadec : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c718d63f7928498c89215d3731c15987 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bccd716db354cbbbbe124dd00fbf6bb : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/204c6244974b43359f3f4d3518d33e8e : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48ccecfddcec47f58c0d8a18ee2fdc9a : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c414ed82c24185ae561f3567374241 : SUCCESS in 4m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9a40b29ef2fa408584ccead2fb5b5cc7 : FAILURE in 2m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb6b0d29ffa448349b7d48e3082adf54 : SUCCESS in 4m 44s\n- searchlight-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b9bfb3e0d9542f0bce2c8f34f3de6f7 : FAILURE in 28m 46s","accounts_in_message":[],"_revision_number":6},{"id":"369de4b35c963e9a00043e8bf50239f36ffe26ae","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-10-18 10:03:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5f6a4180e41c3a1dd0ebb57c9a218ebbd001baa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 10:39:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/856e3623cdc0450b85bc289028eae9cb : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/580ccad35a4c465cb8c17013356dca87 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17fdaf41cc6d4044a4f4bbaa279efc9e : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b19587f32fa4d349f87a2b99ea85534 : SUCCESS in 5m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bc9b5a285854a22987b42a25b22ffee : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c096c7ad9c344b2d9edfcc80d5f4ebc5 : SUCCESS in 5m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d207b1599fd44be592fe9d88f5433f1c : FAILURE in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b946d21c99cc403193d0fba1b4a85caa : SUCCESS in 4m 42s\n- searchlight-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26568d22e34c434fa92a38843b5698bd : FAILURE in 29m 36s","accounts_in_message":[],"_revision_number":7},{"id":"3cf149632158df0560861b9ce43ff5cdbb8ccfe8","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-10-28 16:00:26.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4d66c015af54b43174cb0e0e2f35c2a943a54aff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 16:31:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1e3c051f3c4ba388a0ddfac4202261 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d756dc11b86c447fa032913e95b86ffb : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cc4ad6fc90d49e6b1fcda8bccf9f278 : SUCCESS in 7m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe892463233e45adbd1e13eb6b8e1bb1 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11ebc19771e4d57aba63a8fbcc97a76 : SUCCESS in 5m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf25248c5f954395af6cb6a2c5498f4c : SUCCESS in 2m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b2e1059049d47f88db501fb97485db6 : SUCCESS in 4m 58s\n- searchlight-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ba6d8c0001a4b1490ff946cbaa39f26 : FAILURE in 29m 12s","accounts_in_message":[],"_revision_number":8},{"id":"aa650623a6242b9d23069859f7fbdb70f3587786","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-10-28 16:34:19.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"9d7b1423577b14a74a52ac850a865b13435f398c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 17:05:17.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/340d223a56ed414aba241942bb7e1e67 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb9ae3b2595f4b45b4d18818d6d59523 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ef82d047b9540a396678bff09b50a8c : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc3e88b4f87a45e2aca4bead91695cbc : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb71ab2c6586491089e6219ff5fc9341 : SUCCESS in 5m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b1161621b8cb4b79b3fffe93adf9c66b : SUCCESS in 3m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1107767725d499ea4976b57bd03ddee : SUCCESS in 5m 10s\n- searchlight-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/749b07bfcf4b400b96c8ccc26db3efd5 : FAILURE in 30m 03s","accounts_in_message":[],"_revision_number":9},{"id":"005cfeb1d8a65dd5e171f7809fbd96735088a0cd","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-11-15 01:32:35.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"091bf34f98ffd02608a8244def2cdbe9ee0001a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 02:05:12.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9239da818d3748fe983f2cc7a3443a0f : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34ce0ed7ea18466a84ab4ab3d42335f2 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2540efd81be64c4ebdec370c0ccc155c : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1a47aa508f04fdb8f3bfa837bc6d1ed : FAILURE in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a261d9c53b684e61a468760751851d3a : SUCCESS in 4m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0a93b3ce6854466ea98e8d3d634448b5 : SUCCESS in 2m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/971822a95b784325b9b6ad499ca13a20 : SUCCESS in 5m 09s\n- searchlight-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92fdb7479ead4931bc4b9a2f1eb9ac87 : FAILURE in 28m 47s","accounts_in_message":[],"_revision_number":9},{"id":"99cc97ea7cee38cd5b11e4968098b0f1a8339026","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-04-02 01:46:52.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"01abd2e764b0b7e8a23791ad1d56795d771b96ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 02:19:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7905828013fa49dea99e8d3d3800a840 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7dd7b22a11b4b5088e238ca145151da : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd1e794e22004fb2a2fecbfeafcde71a : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2873d7ca74db4ca8851df52f4beac6d1 : FAILURE in 5m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/717f83b71f5a431ca1cbbd66374bef1f : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea9d01e355e4a99984c309e1d60aa07 : SUCCESS in 6m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/668f27a82fdc49a4835e19ec15daf483 : SUCCESS in 2m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31cab04010a246e8a0fc0f609c7691f2 : SUCCESS in 5m 00s\n- searchlight-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/485774ddeeff42b1aca0e56245758a89 : FAILURE in 27m 48s","accounts_in_message":[],"_revision_number":10},{"id":"58776d733ba53f6e500627348c40035f5b7a9074","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-07 18:30:54.000000000","message":"Abandoned\n\nProject Retirement: http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018637.html","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"09ac164b1aea1db50a02a39504163744ee083007","revisions":{"9f1dcd854a9935516470e20709a54f3b0720106b":{"kind":"REWORK","_number":1,"created":"2019-08-25 06:23:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/91/678391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/91/678391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/91/678391/1"}}},"commit":{"parents":[{"commit":"ec7baccfc4fbf1726d17f763d315604f2f242ae7","subject":"Update api-ref location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/ec7baccfc4fbf1726d17f763d315604f2f242ae7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:21:31.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:23:15.000000000","tz":0},"subject":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027searchlight-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027searchlight-tempest-ipv6-only\u0027 will install searchlight.\n\nStory: #2005477\nTask: #35928\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: I8ee0c999cb7563df0f6f102929dc25f36af623cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/9f1dcd854a9935516470e20709a54f3b0720106b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/9f1dcd854a9935516470e20709a54f3b0720106b"}]},"branch":"refs/heads/master"},"96a7bed35ce185e57d2c75ef0d7d98f68d8ef441":{"kind":"REWORK","_number":2,"created":"2019-08-25 06:32:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/91/678391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/91/678391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/91/678391/2"}}},"commit":{"parents":[{"commit":"ec7baccfc4fbf1726d17f763d315604f2f242ae7","subject":"Update api-ref location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/ec7baccfc4fbf1726d17f763d315604f2f242ae7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:21:31.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:32:12.000000000","tz":0},"subject":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027searchlight-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027searchlight-tempest-ipv6-only\u0027 will install searchlight.\n\nStory: #2005477\nTask: #35928\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: I8ee0c999cb7563df0f6f102929dc25f36af623cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/96a7bed35ce185e57d2c75ef0d7d98f68d8ef441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/96a7bed35ce185e57d2c75ef0d7d98f68d8ef441"}]},"branch":"refs/heads/master"},"04f2846dabf25efc325e032e69e46f8f4523b8a9":{"kind":"REWORK","_number":3,"created":"2019-08-25 07:21:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/91/678391/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/91/678391/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/91/678391/3"}}},"commit":{"parents":[{"commit":"ec7baccfc4fbf1726d17f763d315604f2f242ae7","subject":"Update api-ref location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/ec7baccfc4fbf1726d17f763d315604f2f242ae7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:21:31.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 07:21:32.000000000","tz":0},"subject":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027searchlight-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027searchlight-tempest-ipv6-only\u0027 will install searchlight.\n\nStory: #2005477\nTask: #35928\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: I8ee0c999cb7563df0f6f102929dc25f36af623cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/04f2846dabf25efc325e032e69e46f8f4523b8a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/04f2846dabf25efc325e032e69e46f8f4523b8a9"}]},"branch":"refs/heads/master"},"42fe259949508143b927a32ff958f1d05fb157f5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-18 02:48:49.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/91/678391/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/91/678391/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/91/678391/4"}}},"commit":{"parents":[{"commit":"c8738b8ae188a11780ff22049ea6dca4a8a75d11","subject":"Add PDF generation support for Searchlight","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/c8738b8ae188a11780ff22049ea6dca4a8a75d11"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:21:31.000000000","tz":0},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-10-18 02:48:49.000000000","tz":0},"subject":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027searchlight-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027searchlight-tempest-ipv6-only\u0027 will install searchlight.\n\nStory: #2005477\nTask: #35928\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: I8ee0c999cb7563df0f6f102929dc25f36af623cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/42fe259949508143b927a32ff958f1d05fb157f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/42fe259949508143b927a32ff958f1d05fb157f5"}]},"branch":"refs/heads/master"},"9419bd311e2f17b7387ac7cfcdd4980bcb57019b":{"kind":"REWORK","_number":5,"created":"2019-10-18 02:51:46.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/91/678391/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/91/678391/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/91/678391/5"}}},"commit":{"parents":[{"commit":"c8738b8ae188a11780ff22049ea6dca4a8a75d11","subject":"Add PDF generation support for Searchlight","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/c8738b8ae188a11780ff22049ea6dca4a8a75d11"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:21:31.000000000","tz":0},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-10-18 02:51:19.000000000","tz":540},"subject":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027searchlight-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027searchlight-tempest-ipv6-only\u0027 will install searchlight.\n\nStory: #2005477\nTask: #35928\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: I8ee0c999cb7563df0f6f102929dc25f36af623cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/9419bd311e2f17b7387ac7cfcdd4980bcb57019b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/9419bd311e2f17b7387ac7cfcdd4980bcb57019b"}]},"branch":"refs/heads/master"},"f6ffd36611f0465323c43697844c80b59d4b6fc5":{"kind":"REWORK","_number":6,"created":"2019-10-18 03:01:54.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/91/678391/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/91/678391/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/91/678391/6"}}},"commit":{"parents":[{"commit":"c8738b8ae188a11780ff22049ea6dca4a8a75d11","subject":"Add PDF generation support for Searchlight","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/c8738b8ae188a11780ff22049ea6dca4a8a75d11"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:21:31.000000000","tz":0},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-10-18 03:01:37.000000000","tz":540},"subject":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027searchlight-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027searchlight-tempest-ipv6-only\u0027 will install searchlight.\n\nStory: #2005477\nTask: #35928\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: I8ee0c999cb7563df0f6f102929dc25f36af623cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/f6ffd36611f0465323c43697844c80b59d4b6fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/f6ffd36611f0465323c43697844c80b59d4b6fc5"}]},"branch":"refs/heads/master"},"9c544145c9bbe99231d7832368fe7857eccbb4fc":{"kind":"REWORK","_number":7,"created":"2019-10-18 10:03:24.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/91/678391/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/91/678391/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/91/678391/7"}}},"commit":{"parents":[{"commit":"c8738b8ae188a11780ff22049ea6dca4a8a75d11","subject":"Add PDF generation support for Searchlight","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/c8738b8ae188a11780ff22049ea6dca4a8a75d11"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:21:31.000000000","tz":0},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-10-18 10:02:49.000000000","tz":540},"subject":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027searchlight-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027searchlight-tempest-ipv6-only\u0027 will install searchlight.\n\nStory: #2005477\nTask: #35928\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: I8ee0c999cb7563df0f6f102929dc25f36af623cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/9c544145c9bbe99231d7832368fe7857eccbb4fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/9c544145c9bbe99231d7832368fe7857eccbb4fc"}]},"branch":"refs/heads/master"},"5aaa9efcd2c1cdee3186257fd03a40f94f694820":{"kind":"REWORK","_number":8,"created":"2019-10-28 16:00:26.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/91/678391/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/91/678391/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/91/678391/8"}}},"commit":{"parents":[{"commit":"c8738b8ae188a11780ff22049ea6dca4a8a75d11","subject":"Add PDF generation support for Searchlight","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/c8738b8ae188a11780ff22049ea6dca4a8a75d11"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:21:31.000000000","tz":0},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-10-28 16:00:09.000000000","tz":540},"subject":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027searchlight-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027searchlight-tempest-ipv6-only\u0027 will install searchlight.\n\nStory: #2005477\nTask: #35928\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: I8ee0c999cb7563df0f6f102929dc25f36af623cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/5aaa9efcd2c1cdee3186257fd03a40f94f694820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/5aaa9efcd2c1cdee3186257fd03a40f94f694820"}]},"branch":"refs/heads/master"},"3461e6b8891238f72d387bf2f06c3777a940d20f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-10-28 16:34:19.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/91/678391/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/91/678391/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/91/678391/9"}}},"commit":{"parents":[{"commit":"fe5609304f8721356212eeae049abf87b6484961","subject":"Stop testing Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/fe5609304f8721356212eeae049abf87b6484961"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:21:31.000000000","tz":0},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-10-28 16:34:19.000000000","tz":0},"subject":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027searchlight-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027searchlight-tempest-ipv6-only\u0027 will install searchlight.\n\nStory: #2005477\nTask: #35928\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: I8ee0c999cb7563df0f6f102929dc25f36af623cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/3461e6b8891238f72d387bf2f06c3777a940d20f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/3461e6b8891238f72d387bf2f06c3777a940d20f"}]},"branch":"refs/heads/master"},"09ac164b1aea1db50a02a39504163744ee083007":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-04-02 01:46:52.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/91/678391/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/91/678391/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/678391/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/91/678391/10"}}},"commit":{"parents":[{"commit":"afcc817945fb9af4ca027f4f5fe545d382b4a117","subject":"Remove remaining use of mox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/afcc817945fb9af4ca027f4f5fe545d382b4a117"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-25 06:21:31.000000000","tz":0},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2020-04-02 01:46:52.000000000","tz":0},"subject":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job","message":"[train][goal] Define new \u0027searchlight-tempest-ipv6-only\u0027 job\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027searchlight-tempest-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027searchlight-tempest-ipv6-only\u0027 will install searchlight.\n\nStory: #2005477\nTask: #35928\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: I8ee0c999cb7563df0f6f102929dc25f36af623cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/09ac164b1aea1db50a02a39504163744ee083007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/09ac164b1aea1db50a02a39504163744ee083007"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
