)]}'
{"id":"openstack%2Fsearchlight~694763","triplet_id":"openstack%2Fsearchlight~master~I6b1d709f5d0e4b43e371c4bf856a250a27834e03","project":"openstack/searchlight","branch":"master","hashtags":[],"change_id":"I6b1d709f5d0e4b43e371c4bf856a250a27834e03","subject":"Add option to support python3 in devstack","status":"MERGED","created":"2019-11-18 13:05:50.000000000","updated":"2019-11-18 15:01:11.000000000","submitted":"2019-11-18 14:59:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694763-1574089176957-ded4c5e8","meta_rev_id":"ab165d2c55e6876d1e1b1b81ac2b87d1cdcb0f98","_number":694763,"virtual_id_number":694763,"owner":{"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-18 13:06:01.000000000","_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 14:59:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":2,"date":"2019-11-18 14:49:56.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2019-11-18 14:49:56.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 13:06:01.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-11-18 14:49:56.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-11-18 14:59:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f5295184876f11287ca3650d6b93fdbde0b7f3a","author":{"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},"date":"2019-11-18 13:05:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"650e58c7876c201e9717d2e15617b6072e3c5679","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-11-18 13:06:01.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"b61d2259a2c80560c4150cdcba4d0d66d5ee69a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 13:14:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18f97a720f4a41e6af89f2c8e193ed77 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f44b49eee69e4ed4951dbd6f2eac8605 : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17afb617ce7849f48c6291454e5a802f : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e61a5c84696a434c8e61dd59191d808c : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4bf4a50f024ca3960aedadb585049e : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/009bf571f85444ed9c368f0640ee70ee : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"cda2393c2bf55d231a757fc61a956c2204c396c0","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-11-18 14:49:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"6bbef0df5b20a0e5c73636ef5cbca11c828310d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 14:50:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64863bbad803a12d447d96ab8ab51ec94421b0cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 14:59:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dcd7eb8759cd24702f619a8c349e121bec6429de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 14:59:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57fbe3faa8841258699c1c4b9e9ff0b : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b68f1822b8347829da3d1e18bdac2f1 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c6e9acda05f4bdeac9f1aa9b13d8570 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f170a795573d4be8a8da73d49ed9c666 : SUCCESS in 7m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb710254744e4c7597d420460d832be6 : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ab165d2c55e6876d1e1b1b81ac2b87d1cdcb0f98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 15:01:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c6278bd7884bd1afdb022bcd1a38e5 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f2b2aa3c849c3fe265c43e8587a7b57bfe73e32","revisions":{"2f2b2aa3c849c3fe265c43e8587a7b57bfe73e32":{"kind":"REWORK","_number":1,"created":"2019-11-18 13:05:50.000000000","uploader":{"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},"ref":"refs/changes/63/694763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/63/694763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/63/694763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/63/694763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/63/694763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/63/694763/1"}}},"commit":{"parents":[{"commit":"7361075238ecb8ee60d250c5228d8af74a821695","subject":"Remove image_urls of CirrOS in local.conf.example","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/7361075238ecb8ee60d250c5228d8af74a821695"}]}],"author":{"name":"Shinhocheol","email":"shingoon7@gmail.com","date":"2019-11-18 12:57:49.000000000","tz":540},"committer":{"name":"Shinhocheol","email":"shingoon7@gmail.com","date":"2019-11-18 13:05:38.000000000","tz":540},"subject":"Add option to support python3 in devstack","message":"Add option to support python3 in devstack\n\nBy default stack.sh uses python2.\nYou can ignore this, so devstack runs python3.\nSo in local.conf.example you can use python3 for options\nUSE_PYTHON3 \u003d True has been added.\n\nChange-Id: I6b1d709f5d0e4b43e371c4bf856a250a27834e03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/2f2b2aa3c849c3fe265c43e8587a7b57bfe73e32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/2f2b2aa3c849c3fe265c43e8587a7b57bfe73e32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
