)]}'
{"id":"openstack%2Fsearchlight~700350","triplet_id":"openstack%2Fsearchlight~master~I6d7b469b45bb4c5e7f5e0d793cc76941d76167df","project":"openstack/searchlight","branch":"master","hashtags":[],"change_id":"I6d7b469b45bb4c5e7f5e0d793cc76941d76167df","subject":"tox: Trivial cleanup","status":"MERGED","created":"2019-12-22 16:10:54.000000000","updated":"2020-02-05 13:51:16.000000000","submitted":"2020-02-05 13:48:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700350-1580910508082-71032318","meta_rev_id":"f7c36f1d7ae2acf9db65c0360a68a76ff8883fa3","_number":700350,"virtual_id_number":700350,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-07 01:27:16.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 13:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-02-05 13:24:04.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2020-02-05 13:24:04.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-02 05:41:43.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2020-02-05 13:24:04.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-02-05 13:48:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4587bf9f841d1f51219ad27f911104280da1cfe8","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-22 16:10:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53fa487f23ecbc019a025c5241d8590132ebf52a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 16:18:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2a421fa7da9d462a9ca792c7fb03c086 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5441a2b857224e4e95895c71c090fd45 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d1752317f7c40549d26a8c544930a34 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aee7bb42c6974073a315ab13f54caf32 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f47a2be6af7745d7be1f4174775588f7 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b13b262bec8245f7b3f135add1f06641 : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/413a1aa7d86d40f58492a33e1d0691fb : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd83e30bc1a5440d96b77d25aaa07346 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2b19103e52010635e2822d530b447a65d9754690","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2020-01-02 05:41:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"23f33024fa6c3cbd77b507fde191936d2bddb4d7","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-01-07 01:11:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18b3b0faa135e48affa88ca0e7a7759a05666f6a","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-01-07 01:12:17.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"7a66236b9b4cce970c6cd4b5db7862a60d68ffce","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-01-07 01:26:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3aa646d24dc815d2f3e2aa4e65a9846b1ffa4447","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-01-07 01:27:16.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"203dcd7e23c482795f4db59a42e657bae320d5b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 01:38:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4589d7e2c81042f693638ec1669d3f0e : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd720c5201b54e87bd68034cbb13c060 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88b092cacf96499483b64631740a4fdb : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1250f0d428e7431baabb43fe23adf2cb : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfc68cdaa7284cd7bc8d6625bc3ec125 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ebda879ee6f42519e235e9e119898ed : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/779c8ca9c7f743d2be0c7ed57ba9985d : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/119e005cd9894c1db12ccd6deca456f0 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a7f949bb0c67024e268240b65c9b1cec9da0dd70","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-02-05 13:24:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":2},{"id":"1726d6db03ba666bb26e1267460c247cb186fffc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:24:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e8b6c94028e2dd233ce8d9c6dbf24e44c6ef89d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:48:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/01d6b135231844f587acbf2252052b7f : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99eb46fa8fdd47b6a163820aedf1e05c : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19f4f4bcb0a74d0ea2c3d7f98916a4c9 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9caa6f2c37a74da78ed624117ff5bac9 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d3312399f8432a815254c144c01cf8 : SUCCESS in 7m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18b4c026f44a4cce9a319052770e8369 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a316eef5ccc4c5c95f73b1e2c41f0ed : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"386ed38a6a4796e9c2c346e2b75a4f75ae3cd3ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:48:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f7c36f1d7ae2acf9db65c0360a68a76ff8883fa3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:51:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c4d2b88f5e7443d5b7883d25b1a13b00 : SUCCESS in 1m 22s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ddf33b142d45d284d28faf9ae92ceb : SUCCESS in 2m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a016ed18354b4f2e9ebbcfce639440a6 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5a7674a89cba09b68fd88e5c97cf4d2e1a56bed","revisions":{"b8b6e4289a6e4366b55bb804625207d81f24a87e":{"kind":"REWORK","_number":1,"created":"2019-12-22 16:10:54.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/50/700350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/50/700350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/50/700350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/50/700350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/50/700350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/50/700350/1"}}},"commit":{"parents":[{"commit":"2f2b2aa3c849c3fe265c43e8587a7b57bfe73e32","subject":"Add option to support python3 in devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/2f2b2aa3c849c3fe265c43e8587a7b57bfe73e32"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 16:10:29.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 16:10:29.000000000","tz":480},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nmove \u0027basepython\u0027 to the top-level \u0027testenv\u0027\n\nChange-Id: I6d7b469b45bb4c5e7f5e0d793cc76941d76167df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/b8b6e4289a6e4366b55bb804625207d81f24a87e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/b8b6e4289a6e4366b55bb804625207d81f24a87e"}]},"branch":"refs/heads/master"},"e5a7674a89cba09b68fd88e5c97cf4d2e1a56bed":{"kind":"REWORK","_number":2,"created":"2020-01-07 01:26:45.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/50/700350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/50/700350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/50/700350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/50/700350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/50/700350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/50/700350/2"}}},"commit":{"parents":[{"commit":"2f2b2aa3c849c3fe265c43e8587a7b57bfe73e32","subject":"Add option to support python3 in devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/2f2b2aa3c849c3fe265c43e8587a7b57bfe73e32"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 16:10:29.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2020-01-07 01:26:39.000000000","tz":0},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nmove \u0027basepython\u0027 to the top-level \u0027testenv\u0027\n\nChange-Id: I6d7b469b45bb4c5e7f5e0d793cc76941d76167df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/e5a7674a89cba09b68fd88e5c97cf4d2e1a56bed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/e5a7674a89cba09b68fd88e5c97cf4d2e1a56bed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
