)]}'
{"id":"openstack%2Fsearchlight~709791","triplet_id":"openstack%2Fsearchlight~master~I905b03116d7d7e856581532fda104fe08c4e9f35","project":"openstack/searchlight","branch":"master","topic":"remove-mox","hashtags":[],"change_id":"I905b03116d7d7e856581532fda104fe08c4e9f35","subject":"Remove remaining use of mox","status":"MERGED","created":"2020-02-25 17:06:24.000000000","updated":"2020-02-26 01:48:13.000000000","submitted":"2020-02-26 01:45:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709791-1582681547698-31ef7618","meta_rev_id":"345aa0b5d79e6395a1e33d1e824f0ed008480e66","_number":709791,"virtual_id_number":709791,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 01:45:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-25 19:48:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-26 01:34:16.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-26 01:34:16.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 19:48:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-26 01:34:16.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-02-26 01:45:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a46997175ecc8fbb4577f1f296328d3d7cb41cc4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-25 17:06:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2767e468ff669bd7345b68864269d7e67ed33010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 17:22:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f12afa437a6047e9b77953876cb91a1b : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27eee78257f649cb8e406191a530c3c2 : SUCCESS in 10m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dddb5c9d924943a0afd0c634d6d02dbb : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85ec149170614dd59b5abf7c472b9b03 : FAILURE in 9m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35c10a20aa7041a79211a1472c0fae98 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa9da379f1b548debb48561a582a0472 : SUCCESS in 7m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b594d49f76e14d0481be73dc76850967 : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":1},{"id":"384d819a2a5a897f751747eac46974c7222647ae","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-25 19:48:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"874fee4b26fe28b5c80a8b2a6fa9f5a6998345fc","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-02-26 01:34:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"561cbd6c509dbd1004978ff9e26ef77309202b91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:34:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6bc559d32b909bafc6f9afbdf8076dfc406a34d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:45:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8d8ff5d23a625c6ed4d8d1c1421e270098a723c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:45:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/294a85f7280943b5bd28fb0105718c11 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d95a17b82ad46669a7e48e77a42fcd5 : SUCCESS in 9m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9d475564ef14f6ebf3da92168e67b30 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34caa39cb3e54e0f87d0177317bd0405 : SUCCESS in 7m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf356ac1b2b845c0a0eb45a6cbcd901a : SUCCESS in 9m 28s","accounts_in_message":[],"_revision_number":1},{"id":"eb2e608bcd1f2673b8e9a54f39e8e5db634874d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:48:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c8b98d509ec462393d84bd7ddd8168b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"afcc817945fb9af4ca027f4f5fe545d382b4a117","revisions":{"afcc817945fb9af4ca027f4f5fe545d382b4a117":{"kind":"REWORK","_number":1,"created":"2020-02-25 17:06:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/91/709791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/91/709791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/709791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/709791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/91/709791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/91/709791/1"}}},"commit":{"parents":[{"commit":"09738a8ab9db09da180a5bd9abc3a7fa6648b017","subject":"tox: Set \u0027ignore_basepython_conflict\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/09738a8ab9db09da180a5bd9abc3a7fa6648b017"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-02-25 17:05:01.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-02-25 17:06:11.000000000","tz":0},"subject":"Remove remaining use of mox","message":"Remove remaining use of mox\n\nThis is pretty trivial but kills mox usage in this project.\n\nChange-Id: I905b03116d7d7e856581532fda104fe08c4e9f35\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/afcc817945fb9af4ca027f4f5fe545d382b4a117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/afcc817945fb9af4ca027f4f5fe545d382b4a117"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
