)]}'
{"id":"openstack%2Fsearchlight~717658","triplet_id":"openstack%2Fsearchlight~master~Ia561a40708efb4346146ab0456adc1a1f1565b47","project":"openstack/searchlight","branch":"master","hashtags":[],"change_id":"Ia561a40708efb4346146ab0456adc1a1f1565b47","subject":"Use unittest.mock instead of third party mock","status":"MERGED","created":"2020-04-06 04:48:35.000000000","updated":"2020-04-07 17:13:56.000000000","submitted":"2020-04-07 17:10:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717658-1586279436794-25f1bb97","meta_rev_id":"2796e14ac2f59f7ae3d6238477a908f398913001","_number":717658,"virtual_id_number":717658,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-06 05:14:53.000000000","_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 17:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2020-04-07 16:56:17.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2020-04-07 16:56:17.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 16:56:17.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-04-07 17:10:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7eaa95351a1584ad86ace4a93ba7c027b6ea90f8","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-04-06 04:48:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f2f59d25fcc6a63d06778cf8a2a57c53c23c7b1","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-04-06 04:58:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b7780595431b79c1187fae55d5633f8ea43d2ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 05:07:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/96ffa11451b74831a804684a6a768529 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3797fe37a074896b725b3a33f633f91 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/baddb2b26118446cb15918121f613d5b : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7b6b7e9e036499b9f383da2b9fb2116 : FAILURE in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fff8b197b6e4441fb7480d78681e2a4a : FAILURE in 4m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ab22b85574e4b1dabb86b5b5627f675 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87497af1a23c4393a4ed3af61a4f55d0 : SUCCESS in 4m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/66c2ed9b63fc4b939e5cc1a18ca853ef : SUCCESS in 2m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0dbc98f0d3e4b78abf4a28430dcd63f : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c813f503ba145aca3f8e6610c9e55b6 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"f68ae7ac5e6386178ea055999dca44a6bb193f96","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-04-06 05:14:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fa4ec02b99d2763734fe68051ef1f3df03f87812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 05:22:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/022bede79184496a834da29bee551ee4 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e9a13e3caf4adcbb6316a07ca22b0a : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d164ff44e97444b8100605aff38ae8d : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13ed95d709e3455c941dbb824b3c6bdc : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1325f88d3e45435487a4e7876e084f32 : FAILURE in 5m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f3640f0542f4ee198837c9b5b421a9f : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f53dfa38f5945fa8b364c0aeb31f004 : SUCCESS in 5m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f6e2914d372a40268d3620c70b3b2318 : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/352d9a8541444ac595d56e5296d96708 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92e8364da44a4a7397b5a349dc05c289 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"54dab3e0956bdfeb8253d5fb2496654c8d21317c","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-04-07 16:56:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":2},{"id":"f17a776a23f9bbfccb06a420ae0102774413c50a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 16:56:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b2e048853e96ebfd174564be3f7a9fd2fd9d6801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 17:10:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3cb3843878634a10af7dfed96177c22d : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5c28d2d409247d39727e2ae69dbddde : SUCCESS in 8m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dccd2ae26d0434f8608838fed9f5992 : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ae521c274354adf819d70f1d6b06868 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8baadea1614428b1bdfb9817547cea : SUCCESS in 8m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fce9ab6efd0c43bba29dbf1274a3333a : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f61e7b9d74334e6d825ea0b1970c6b17 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a817e15ed884180b69c0f6247a69c3d : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"cf2bca7e53c0ddcd9fbe86dccf849dd336ffcfe0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 17:10:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2796e14ac2f59f7ae3d6238477a908f398913001","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 17:13:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/44fe7ba283d44ac9ae134543c3821b63 : SUCCESS in 2m 28s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ede8745ebc43cd9acc9c13236b25e9 : SUCCESS in 2m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1242dc9eea724d2783d2dcef03cbde97 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cddffe6af3d816ac862819c891944693390fac39","revisions":{"a684c2631a5d7d1c928f0a44064a77e88bc7d087":{"kind":"REWORK","_number":1,"created":"2020-04-06 04:48:35.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/58/717658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/58/717658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/58/717658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/58/717658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/58/717658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/58/717658/1"}}},"commit":{"parents":[{"commit":"0d9a00306025bccff7dbcc63ab08cb4f2f863f22","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/0d9a00306025bccff7dbcc63ab08cb4f2f863f22"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-04-06 04:47:42.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-04-06 04:48:27.000000000","tz":480},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: Ia561a40708efb4346146ab0456adc1a1f1565b47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/a684c2631a5d7d1c928f0a44064a77e88bc7d087"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/a684c2631a5d7d1c928f0a44064a77e88bc7d087"}]},"branch":"refs/heads/master"},"cddffe6af3d816ac862819c891944693390fac39":{"kind":"REWORK","_number":2,"created":"2020-04-06 04:58:00.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/58/717658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/58/717658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/58/717658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/58/717658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/58/717658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/58/717658/2"}}},"commit":{"parents":[{"commit":"0d9a00306025bccff7dbcc63ab08cb4f2f863f22","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/0d9a00306025bccff7dbcc63ab08cb4f2f863f22"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-04-06 04:47:42.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-04-06 04:57:43.000000000","tz":480},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: Ia561a40708efb4346146ab0456adc1a1f1565b47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/cddffe6af3d816ac862819c891944693390fac39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/cddffe6af3d816ac862819c891944693390fac39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
