)]}'
{"id":"openstack%2Fsearchlight~721869","triplet_id":"openstack%2Fsearchlight~master~I76c3a7e63a63ce08a03a6b0d3d3ae34f9f038f59","project":"openstack/searchlight","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"I76c3a7e63a63ce08a03a6b0d3d3ae34f9f038f59","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-04-22 12:54:15.000000000","updated":"2020-06-09 14:08:26.000000000","submitted":"2020-06-09 14:04:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721869-1591711471382-00e32a87","meta_rev_id":"33c257597ec6fb8a27b4d32f7ed6cc3c7e80acc8","_number":721869,"virtual_id_number":721869,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 14:04:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},{"value":2,"date":"2020-06-09 13:46:42.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"date":"2020-06-09 09:55:08.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-09 09:55:08.000000000","_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},{"value":1,"date":"2020-06-09 13:46:42.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 09:55:08.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-09 09:55:08.000000000","updated_by":{"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},"reviewer":{"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},"state":"REVIEWER"},{"updated":"2020-06-09 13:46:42.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-06-09 14:04:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c1569e7a6f1deeb41cbdd492eab69057718ccb9","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-22 12:54:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9ec54d91bc23f2ea4b2fbd77be2f8f771a73b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 13:23:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/476348afb783453ea51b191241a8928c : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c397c61279ea4c5bbaea9cc0261d2db5 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c3fb2564cb9435ab0af57223118df6a : FAILURE in 7m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5369bd3437434bbf961014dc8348c867 : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce4364e141e0470b905b5dd9f73e2341 : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44b001f99cb54460a1335c2025b7e85d : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ee07b4984287fa1bc8d61b5aebb29ad04285c96d","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-05-06 11:50:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3ad5fa022428900fe3f783b4799488d2097aa180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 12:00:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9041ddf093634a52868afad351fad781 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61be6ad419c9417e861da977291c9a29 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c475a436a1ed487aa5af3d43b0024dbb : FAILURE in 8m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5b82d9b7264402ab2a18b631c2fa89f : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89a8479e277d40a6bd8ee1f35006877f : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1974655fe4f5446c8841193607576da3 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":2},{"id":"36a2d83ada693d3b2d4a907977b9edc5664d2ab2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 09:47:17.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"498af92a3434eb6d6359eefd7216d25c62b32b16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 10:16:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e05b62ea66c46cba9e15fb2803d6b62 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44d8dafe61ac4a1cbc3c429a10cb211c : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bee383c89093443a9b6cb2b9e69ed9bf : SUCCESS in 9m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0018aa24bbbb4804b6b9e6350dd9a549 : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c45556517b154541adc494d2d50248d3 : SUCCESS in 9m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a7c6e617abb4e82897b6558ebebc7ba : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":3},{"id":"f9531017707b04e5d45c1e452921db5113bf13a5","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-06-09 13:46:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"80781c8c2bcbf6669235b06625528d69aee4db68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 13:47:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d3af9f233e3cf046e8fa11da80a3cc1d1684f4fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:04:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0c256da2b5cce2c3ef6a269e9e7dd2eb614a8ab0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:04:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/281be8bc04934d1994541f7894a27a39 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2408a4e06127405894831ca63343e043 : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b112edfbbc34440a75a8d0292b69b22 : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6457f981f7344eeaa8c1d218883bac6 : SUCCESS in 7m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e678dccecd584bdc93ae7ae68694565f : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":3},{"id":"4ca4ececb1cdb89b9d3b87061605aa84c5d3474c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:08:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1244a3b445344c38d8e4c8d6f7bffd4 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6ef065688cc0fc17683d61ce2faac81cf0070326","revisions":{"a61c96c497ae2c57cb286b5548acf1fa94cb591d":{"kind":"REWORK","_number":1,"created":"2020-04-22 12:54:15.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/69/721869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/69/721869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/69/721869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/69/721869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/69/721869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/69/721869/1"}}},"commit":{"parents":[{"commit":"1103aee211adcf521db79f1fdd36d09f1b1d6e43","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/1103aee211adcf521db79f1fdd36d09f1b1d6e43"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 12:54:12.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 12:54:12.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I76c3a7e63a63ce08a03a6b0d3d3ae34f9f038f59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/a61c96c497ae2c57cb286b5548acf1fa94cb591d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/a61c96c497ae2c57cb286b5548acf1fa94cb591d"}]},"branch":"refs/heads/master"},"be276f1c7cd5fc958bdb88164b3a282597c85389":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-06 11:50:21.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/69/721869/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/69/721869/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/69/721869/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/69/721869/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/69/721869/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/69/721869/2"}}},"commit":{"parents":[{"commit":"e276e92057b7e70b726e82594c132ed6cb07f61f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/e276e92057b7e70b726e82594c132ed6cb07f61f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 12:54:12.000000000","tz":0},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2020-05-06 11:50:21.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I76c3a7e63a63ce08a03a6b0d3d3ae34f9f038f59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/be276f1c7cd5fc958bdb88164b3a282597c85389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/be276f1c7cd5fc958bdb88164b3a282597c85389"}]},"branch":"refs/heads/master"},"6ef065688cc0fc17683d61ce2faac81cf0070326":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-09 09:47:17.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/69/721869/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/69/721869/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/69/721869/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/69/721869/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/69/721869/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/69/721869/3"}}},"commit":{"parents":[{"commit":"14798f56b053f8a8819b6f51e26839cc746a3d16","subject":"Fix py38 problem","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/14798f56b053f8a8819b6f51e26839cc746a3d16"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 12:54:12.000000000","tz":0},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-06-09 09:47:17.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I76c3a7e63a63ce08a03a6b0d3d3ae34f9f038f59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/6ef065688cc0fc17683d61ce2faac81cf0070326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/6ef065688cc0fc17683d61ce2faac81cf0070326"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
