)]}'
{"id":"openstack%2Fsearchlight~722905","triplet_id":"openstack%2Fsearchlight~master~I3ed9bd1067d9c6929cd1dda44f501a7c124dc493","project":"openstack/searchlight","branch":"master","topic":"py38-tox","hashtags":[],"change_id":"I3ed9bd1067d9c6929cd1dda44f501a7c124dc493","subject":"Bump default tox env from py37 to py38","status":"MERGED","created":"2020-04-24 15:37:15.000000000","updated":"2020-04-25 01:39:48.000000000","submitted":"2020-04-25 01:38:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722905-1587778695166-16b58591","meta_rev_id":"74febb7f7493d3dc17d61441d8e875f7fc5852b5","_number":722905,"virtual_id_number":722905,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-25 01:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2020-04-25 01:30:11.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2020-04-25 01:30:11.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-25 01:30:11.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-04-25 01:38:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fda9aa4412c1c9128f93ce9b05f61614d10925d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 15:37:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a07f09e98fdd0a7879a486c9c71ceaabb0c01288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 16:19:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0daeb2ac43004c548a586d10cba85a20 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf5a7593bcc740098db7f3671fdd9ec4 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87a8ba31d6664971a290aa82b0f3b8db : SUCCESS in 8m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5e6c4c486a54c97b7f42a6dba92cae2 : SUCCESS in 10m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f248ee45872042538f7f3a07ef7bdf97 : FAILURE in 7m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e116cebb16d4af1a04aa30b01995158 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe057f8151d4c58a727b2f874608a4a : SUCCESS in 8m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83b43df57c0e4e159dd5f5a859ee99f1 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2f23e74563d46cfa8515d8c5e340af9 : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9b0070a6d9383bfbbea6066236f1c424c3bf6375","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-04-25 01:30:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d5f5b826965e0b782547ce9d20423e2dec218d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 01:30:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f96f1fc2c7115f1706afec204e46d33308a2b7ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 01:38:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2794b155ed6e48eba8639b934e075881 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc8b692209ea4d0bae0859a8791cf756 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71f106714bc84e84ae5cbb2c7c88d3b1 : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2719e0e6b55f47609cef2d15dcda521c : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef9bdea1c0c145ffb0797c5995138e14 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58e4c3f5b5564f64a402900127a268e6 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28abfc19e0f546fdb06ae8a9304a69a5 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"72f22131e5a5382eb62afe704ca7ad08c2c964a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 01:38:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"52aded7eee83a37cf41541d8c7db6b67cec0a02e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 01:39:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/cd1c443c3a9e4488ab7bf5fb81661247 : SUCCESS in 1m 05s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/109d65bcc74d425080cdcf9dcc630efb : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c547287cd024134a61428624775f29e : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8c498d2ffd229f53c015ea305f9d41fe289e705","revisions":{"e8c498d2ffd229f53c015ea305f9d41fe289e705":{"kind":"REWORK","_number":1,"created":"2020-04-24 15:37:15.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/05/722905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/05/722905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/05/722905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/05/722905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/05/722905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/05/722905/1"}}},"commit":{"parents":[{"commit":"1103aee211adcf521db79f1fdd36d09f1b1d6e43","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/1103aee211adcf521db79f1fdd36d09f1b1d6e43"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:26:02.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:26:02.000000000","tz":-300},"subject":"Bump default tox env from py37 to py38","message":"Bump default tox env from py37 to py38\n\nPython 3.8 is now our highest level supported python runtime.\nThis updates the default tox target environments to swap out\npy37 for py38 to make sure local development testing is\ncovering this version.\n\nThis does not impact zuul jobs in any way, nor prevent local\ntests against py37. It just changes the default if none is\nexplicitly provided.\n\nChange-Id: I3ed9bd1067d9c6929cd1dda44f501a7c124dc493\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/e8c498d2ffd229f53c015ea305f9d41fe289e705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/e8c498d2ffd229f53c015ea305f9d41fe289e705"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
