)]}'
{"id":"openstack%2Fsearchlight~729655","triplet_id":"openstack%2Fsearchlight~master~I2421d8bcf88633c27c80d620f598365e30e55aeb","project":"openstack/searchlight","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"I2421d8bcf88633c27c80d620f598365e30e55aeb","subject":"Switch to newer openstackdocstheme and reno versions","status":"MERGED","created":"2020-05-20 16:26:28.000000000","updated":"2020-05-31 14:55:08.000000000","submitted":"2020-05-31 14:53:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729655-1590936832925-06e40272","meta_rev_id":"8a80d8c432a777fa912fe0c3856bea1e05d8447c","_number":729655,"virtual_id_number":729655,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-31 14:53:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-31 14:43:30.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-31 14:43:30.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 16:48:13.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-29 09:14:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-31 14:43:30.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-05-31 14:53:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"057ca4de4b1f7a2647883a1a9f298112d66b54c6","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-20 16:26:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0996854dfd7bf1e697c9801a7c6274d1c639f85","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-20 16:28:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"132b54bc80f01e7e3b5a05e02cff7b19d37fa5ec","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-20 16:34:41.000000000","message":"Patch Set 2:\n\n@Andreas \n\nHI Andreas, please help review this PS, thanks","accounts_in_message":[],"_revision_number":2},{"id":"fe2433340b8078121f0553fb6c2fd088aac4f462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 16:47:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/27ee3780e07343a093c726a57afd7e1d : SUCCESS in 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df51ce05f4b41538914d1d6b7ca90e0 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4abcf2d34694990b305ef761879ece7 : SUCCESS in 11m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66c7a065ba724bad89fd74fb7b5a1bb7 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e802de45286e4eb3931cc7ed952aeaa7 : FAILURE in 8m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2657555c05340af93329496d47c2a50 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2584cc087d534f6fb4a68982e5036b86 : SUCCESS in 9m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/45ddca6e2b6345ffa834fed23ee58464 : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ec8aa6695c44fe692901635237c2214 : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b195bda8682b4f7786c8cb7b21eac3b7 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a407dbb133f9809619eac04f253d8838c70c20ab","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-20 16:48:13.000000000","message":"Patch Set 2:\n\nPlease use the same topic that I use - reno-openstackdocstheme to avoid duplicates.","accounts_in_message":[],"_revision_number":2},{"id":"282ea4e1e58c17708fc284510a3dcaaf44332980","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-20 16:48:19.000000000","message":"Topic set to reno-openstackdocstheme","accounts_in_message":[],"_revision_number":2},{"id":"f290f298b4a2df799b09605c773fc964f259ba1b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-29 09:14:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"655aaa8b4ffe0f9a039c93367e62dc3e2abd2d14","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-29 09:39:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e90f0e0434572a7a1f61115b6d292336bc1c1a31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 09:51:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c0fafad2a00b44bbacfda33c9c821dd1 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85e321305f654de1a431bed24385e638 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5a905a9a7b34360821d739dfb1d8637 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/adcf4f97fd09495c95f7f83eaea5adef : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23feddac0b76481e9777548c1e7ac6d7 : FAILURE in 6m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb3c8a200c3c4e06ade5552a224fa64d : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db88816f6aa346b4bca01ce8e628e785 : SUCCESS in 6m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/61c342141f53455baaeeb4a727aa7190 : SUCCESS in 3m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06afffe8db3a45d2a1758fdb7f762606 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f25b87820b3940dba9aafed3823e468e : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f31ce138d09482cac548b8f22b8f7995f5df9fa2","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-05-31 14:43:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":3},{"id":"946fdf4a22b9220bf5f7e12234c67d9f9fc14d07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 14:43:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"223c5c43e3ed38c4429527d2dd2dfd5d100f062d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 14:53:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4fdf14987e82ef9dbc5b3ac820486e2788f5408a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 14:53:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8b1328c131d8427fa2d4db3250df78d6 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcafbe226ba0437f88b31c76ea0e396b : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/795ea8ee6a60414f864b24586fc518ac : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/506d9dd3dcac4fd4a72837b754eddc01 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/749ae23626f94f529d8dc0a712fa6cb9 : SUCCESS in 6m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/85d78e6f0547415fa20494725502d69a : SUCCESS in 2m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/197dae2b698948aea47510f047c693ea : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2ffcf864b47423389dabe7566636742 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":3},{"id":"8a80d8c432a777fa912fe0c3856bea1e05d8447c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 14:55:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/8ae6645d96894359adddbfc84089bd71 : SUCCESS in 58s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c68128b448ea4ae3845ef4198ef53331 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/562a2fd83b3d417f828290e1cfa998ea : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"592bb6ce7dfcd4dcab24293cf9fd4d72f3fd92f4","revisions":{"bcae84260d33f7ce039cc0b6801943418b2bd6a0":{"kind":"REWORK","_number":1,"created":"2020-05-20 16:26:28.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/55/729655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/55/729655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/55/729655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/55/729655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/55/729655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/55/729655/1"}}},"commit":{"parents":[{"commit":"13731494d7c1b0c641cdb8203176cc1c8cf7a1f5","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/13731494d7c1b0c641cdb8203176cc1c8cf7a1f5"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-20 16:25:24.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-20 16:25:24.000000000","tz":480},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.0 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_pdf_link to link to PDF file. Note that\nthe link to the published document only works on docs.openstack.org\nwhere the PDF file is placed in the top-level html directory. The\nsite-preview places the PDF in a pdf directory.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nRemove docs requirements from lower-constraints, they are not needed\nduring install or test but only for docs building.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nChange-Id: I2421d8bcf88633c27c80d620f598365e30e55aeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/bcae84260d33f7ce039cc0b6801943418b2bd6a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/bcae84260d33f7ce039cc0b6801943418b2bd6a0"}]},"branch":"refs/heads/master"},"50bee10d4e1e264e3937da7ad4827dae68ffb787":{"kind":"REWORK","_number":2,"created":"2020-05-20 16:28:25.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/55/729655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/55/729655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/55/729655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/55/729655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/55/729655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/55/729655/2"}}},"commit":{"parents":[{"commit":"13731494d7c1b0c641cdb8203176cc1c8cf7a1f5","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/13731494d7c1b0c641cdb8203176cc1c8cf7a1f5"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-20 16:25:24.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-20 16:27:47.000000000","tz":480},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.0 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_pdf_link to link to PDF file. Note that\nthe link to the published document only works on docs.openstack.org\nwhere the PDF file is placed in the top-level html directory. The\nsite-preview places the PDF in a pdf directory.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nRemove docs requirements from lower-constraints, they are not needed\nduring install or test but only for docs building.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nChange-Id: I2421d8bcf88633c27c80d620f598365e30e55aeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/50bee10d4e1e264e3937da7ad4827dae68ffb787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/50bee10d4e1e264e3937da7ad4827dae68ffb787"}]},"branch":"refs/heads/master"},"592bb6ce7dfcd4dcab24293cf9fd4d72f3fd92f4":{"kind":"REWORK","_number":3,"created":"2020-05-29 09:39:08.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/55/729655/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/55/729655/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/55/729655/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/55/729655/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/55/729655/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/55/729655/3"}}},"commit":{"parents":[{"commit":"13731494d7c1b0c641cdb8203176cc1c8cf7a1f5","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/13731494d7c1b0c641cdb8203176cc1c8cf7a1f5"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-20 16:25:24.000000000","tz":480},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-05-29 09:38:58.000000000","tz":0},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_pdf_link to link to PDF file. Note that\nthe link to the published document only works on docs.openstack.org\nwhere the PDF file is placed in the top-level html directory. The\nsite-preview places the PDF in a pdf directory.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nRemove docs requirements from lower-constraints, they are not needed\nduring install or test but only for docs building.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nChange-Id: I2421d8bcf88633c27c80d620f598365e30e55aeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/592bb6ce7dfcd4dcab24293cf9fd4d72f3fd92f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/592bb6ce7dfcd4dcab24293cf9fd4d72f3fd92f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
