)]}'
{"id":"openstack%2Fsearchlight~732951","triplet_id":"openstack%2Fsearchlight~master~I2a58e8d4c6f4331104474c216c99427db49f1cab","project":"openstack/searchlight","branch":"master","topic":"drop_future_imports","hashtags":[],"change_id":"I2a58e8d4c6f4331104474c216c99427db49f1cab","subject":"Stop to use the __future__ module.","status":"ABANDONED","created":"2020-06-02 18:58:14.000000000","updated":"2020-12-07 18:30:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d18181658d400c75aa12edcc8708899589371ce2","_number":732951,"virtual_id_number":732951,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"date":"2020-07-10 11:22:05.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-09 15:11:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-1,"date":"2020-07-28 16:58:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-07-10 13:44:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 15:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-10 13:44:17.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-07-28 16:58:54.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"c04d1b2997ece3ac7f7738a2ead20591a1889e92","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 18:58:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a81f922376e6f4a3a97d6b1b4ede73d1e39adb8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 19:12:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed91a54ba2da4ca892b34c13316a8fd4 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6830143b124443888cc5e6c5c5fc224 : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ab5897ae30b43de8c250bb9a5dfce26 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eca45661dd324283ba0bc2b89b235abe : FAILURE in 9m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e78385f04ad422f9664daff31b5d60f : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/189c66ff9d7342f38648316bb7d8b9dd : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e48c1a68ebdd4dd0853c6ccd3f2d97a5 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"53050f8bf07465245f775327f38474a1ff75b25a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-09 09:47:22.000000000","message":"Patch Set 1:\n\nHello, do you think is it possible to ask for someone to merge these changes to avoid possible future conflicts and rebases.","accounts_in_message":[],"_revision_number":1},{"id":"b3fda5f627e2aca8a7966eedd9ec974bbf589436","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-07-09 14:31:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"944eb0ba7f00a0babcb081b7cb339ef52bb7054d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 14:31:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9100d83c4c3f0882c40b5f19d1fe3d0a853a036e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 14:45:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d20f04a5ef464d1fb7eb201e4eece5f5 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdeeee146fb44a6593bd02ca07c3286c : FAILURE in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1aaaeb631e8f488bb711a7631431a210 : FAILURE in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d994ccdd8045411388ab7cb0421a1657 : SUCCESS in 12m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/642019bc1272411cb8153970529d79f7 : FAILURE in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f65cc84a4967e02ede6e9cf912c10af682022db4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-09 14:54:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e305fcc7b5b6ffdf6606de3f823978326da89541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 15:11:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c316e15ec42473fa044bd8b8ded02bf : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8553e6be77dc44a5b1a3e774961b1bf8 : FAILURE in 8m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59f165b7edd841da82f162eb24b2b6a4 : FAILURE in 8m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/410ebc9dbbc242198ca5e68be1ec18e6 : FAILURE in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa06286c26645bc8b9f8e0eccbe0727 : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b84747e39db43b1825c1b257c1151f0 : FAILURE in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6b4a92cf732fc2c0a1f88d788cbecfdcb08563cf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-09 16:01:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6da92774c805f5789660e1d2659066e61f09d31f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 16:18:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c58862b3d28447f1a4976493fd78cdec : SUCCESS in 9m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6beee0b524ec44108052dc6b8edfb5de : FAILURE in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32cea9ef71f24422a1360f9bd7a4cb38 : FAILURE in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67fa665bf630441b8903e2faa5cea87b : FAILURE in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c26acf2e32c46dd8cd8dfce95365dce : SUCCESS in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/823841a4d2224b35acd917c4fb560021 : FAILURE in 9m 56s","accounts_in_message":[],"_revision_number":1},{"id":"ef00ba2b42ef0e00a5c809f8273eeadb20bc2d6f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-09 22:54:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b13db65fbf451b32af3cce1f936213023dacc4c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 23:06:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c4cce359b1743d39743aadc5298d129 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b52ee44c542494c9e203b993a035a1c : FAILURE in 11m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaa44dc552624df8a5017c92935df4a8 : FAILURE in 10m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/124323edb7294d11853e7714c3100a6d : FAILURE in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79bf506060904ef59d3981769f4ad649 : SUCCESS in 10m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fab3fcc87b54de98a23ee213b9b0ef1 : FAILURE in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"163c5142b01f629d4465688271c432a18e6bc310","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-10 07:53:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b2086c6b0617150273778bff4651f30aad3b755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 08:06:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20c4b740a6df4ed1b441a8c86b14c0b2 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81fd5f9ad43b41ca88b96b5f5455d231 : FAILURE in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae04324ec9e44da28933daea0111bb39 : FAILURE in 8m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd70aab26779445aad3f7d08693ed6ef : FAILURE in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e319e844194aee899afd3b02583ab7 : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d452944096dd493593c2d95d46baefac : FAILURE in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9cdbef67df52b6c9ddd7c635bdd5f4858bc1f8a3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-10 08:11:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1497666b26d1d4ebdd6230966753c77142ffbd3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 08:31:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e575648a034b27a35d369a963387ca : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d92f0ca44dc4dd699e6783b6c888260 : FAILURE in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54a04cc0dff440c3848adeb9715894c2 : FAILURE in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d28455835471436e875440d60424794e : FAILURE in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7d8f48899e45a688bfdbf544b948b5 : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5694a252ef434ba69572ab3406ef09e1 : FAILURE in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"044ff563f3ba2fbd2b23f8a7f09afdfc67433655","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-10 09:28:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e80ace67523a7ea3359255ad72c531c1722e2571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 09:50:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3300f5a00ccb43fbb32110f5fa2bf26e : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/530a10b9264f482c8c088dc87e7d9f33 : FAILURE in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9303bbb2cefc44828357e1db16ca76e9 : FAILURE in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc4b3d5ebe54423db39ffb643c6e634b : FAILURE in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dbe4ab6628a4b0f863c6c57f290faf5 : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/904077c86c1341ccb0f322f0cc5edccd : FAILURE in 8m 11s","accounts_in_message":[],"_revision_number":1},{"id":"49f6db0e4d7ede4efccd314db7513ca640c32ae3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-10 11:22:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b895886a118186a4da55801f7fd76057a5bcba80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 11:35:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9ad3b05e12b458d9b4b90fe61047428 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76885b3048bb4cae8002880859f29d9a : FAILURE in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b147caeceba54956b00c51e42148f082 : FAILURE in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4541d68398ae496fbf3402ed981ac876 : FAILURE in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/620b980d85e64ceda9db86ecf4932450 : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/940e393e8e8a46e5a78d7484f8cad8bf : FAILURE in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"769685ed68d82fc28ccf86be3f3c48340e7c6a82","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-07-10 13:44:17.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nMust be something has changed recently.\n\nft1.5: searchlight.tests.unit.test_wsgi.RequestTest.test_request_match_language_unexpectedtesttools.testresult.real._StringException: traceback-1: {{{\nTraceback (most recent call last):\n  File \"/home/zuul/src/opendev.org/openstack/searchlight/.tox/lower-constraints/lib/python3.6/site-packages/fixtures/fixture.py\", line 197, in setUp\n    self._setUp()\n  File \"/home/zuul/src/opendev.org/openstack/searchlight/.tox/lower-constraints/lib/python3.6/site-packages/fixtures/_fixtures/monkeypatch.py\", line 143, in _setUp\n    current \u003d __import__(components[0], {}, {})\nModuleNotFoundError: No module named \u0027babel\u0027\n\nDuring handling of the above exception, another exception occurred:\n\nTraceback (most recent call last):\n  File \"/home/zuul/src/opendev.org/openstack/searchlight/.tox/lower-constraints/lib/python3.6/site-packages/fixtures/fixture.py\", line 208, in setUp\n    raise SetupError(details)\nfixtures.fixture.SetupError: {}\n}}}\n\nTraceback (most recent call last):\n  File \"/home/zuul/src/opendev.org/openstack/searchlight/.tox/lower-constraints/lib/python3.6/site-packages/fixtures/fixture.py\", line 197, in setUp\n    self._setUp()\n  File \"/home/zuul/src/opendev.org/openstack/searchlight/.tox/lower-constraints/lib/python3.6/site-packages/fixtures/_fixtures/monkeypatch.py\", line 143, in _setUp\n    current \u003d __import__(components[0], {}, {})\nModuleNotFoundError: No module named \u0027babel\u0027","accounts_in_message":[],"_revision_number":1},{"id":"928631ac1ae983f3ca1f8da84a1447957a9a94f3","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-07-28 16:58:54.000000000","message":"Patch Set 1: Code-Review-1\n\nDuplicate. This change is already merged by https://review.opendev.org/#/c/737139/","accounts_in_message":[],"_revision_number":1},{"id":"d18181658d400c75aa12edcc8708899589371ce2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-07 18:30:36.000000000","message":"Abandoned\n\nProject Retirement: http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018637.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6e2a4bebebaed1c37cfbc998ace992164259a57","revisions":{"d6e2a4bebebaed1c37cfbc998ace992164259a57":{"kind":"REWORK","_number":1,"created":"2020-06-02 18:58:14.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/732951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/51/732951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/51/732951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/51/732951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/51/732951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/51/732951/1"}}},"commit":{"parents":[{"commit":"ea0aa97f212ae70714104c3a0923902fb6536cd8","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/ea0aa97f212ae70714104c3a0923902fb6536cd8"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 18:58:07.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 18:58:07.000000000","tz":120},"subject":"Stop to use the __future__ module.","message":"Stop to use the __future__ module.\n\nThe __future__ module [1] was used in this context to ensure compatibility\nbetween python 2 and python 3.\n\nWe previously dropped the support of python 2.7 [2] and now we only support\npython 3 so we don\u0027t need to continue to use this module and the imports\nlisted below.\n\nImports commonly used and their related PEPs:\n- `division` is related to PEP 238 [3]\n- `print_function` is related to PEP 3105 [4]\n- `unicode_literals` is related to PEP 3112 [5]\n- `with_statement` is related to PEP 343 [6]\n- `absolute_import` is related to PEP 328 [7]\n\n[1] https://docs.python.org/3/library/__future__.html\n[2] https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n[3] https://www.python.org/dev/peps/pep-0238\n[4] https://www.python.org/dev/peps/pep-3105\n[5] https://www.python.org/dev/peps/pep-3112\n[6] https://www.python.org/dev/peps/pep-0343\n[7] https://www.python.org/dev/peps/pep-0328\n\nChange-Id: I2a58e8d4c6f4331104474c216c99427db49f1cab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/d6e2a4bebebaed1c37cfbc998ace992164259a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/d6e2a4bebebaed1c37cfbc998ace992164259a57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
