)]}'
{"id":"openstack%2Fsearchlight~733508","triplet_id":"openstack%2Fsearchlight~master~I4267071cdfcb6319ed02e355a9c761fce58e5383","project":"openstack/searchlight","branch":"master","topic":"reset_conf","hashtags":[],"change_id":"I4267071cdfcb6319ed02e355a9c761fce58e5383","subject":"Always reset CONF when starting the wsgi app","status":"MERGED","created":"2020-06-04 09:32:36.000000000","updated":"2020-06-04 15:18:25.000000000","submitted":"2020-06-04 15:16:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733508-1591283788156-0bd3b327","meta_rev_id":"a1a0fb72ed4372411a2d45116fd84710a8dc49e2","_number":733508,"virtual_id_number":733508,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 15:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2020-06-04 15:02:50.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2020-06-04 15:02:50.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 15:02:50.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-06-04 15:16:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73d547d21004ed512b440316217c2a56ea058e8f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-04 09:32:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96dcd7e4576116a688fc6114ea096ab8d3929986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:45:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/133d410838dc446997e34b457915e7db : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/736f98750e5d4e4a9a1eee7cd08a58a2 : SUCCESS in 7m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2c130d3059241d993171cef7d08544f : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f355adf115540d190fbd9c300a52095 : FAILURE in 5m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a1f41117ee4487e9e3aada979eb47ae : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7d95d34daef4a689c6086492a3d0bf8 : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff17d1f6310a408ab2c406f4e67836e5 : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2b60aefa155247f4c37ae7128cfe6f916eaf1aaf","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-06-04 15:02:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM. Thanks","accounts_in_message":[],"_revision_number":1},{"id":"a0cdfea855a2da2d9b42ff39c04466fd839198de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 15:03:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"910c582b1e356e4791fc2a75a9e7140ac7d7f8f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 15:16:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f8b4d08819c4ac5b28077b71804d918 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/508e81a576404fa48bd7d9f5c6825e83 : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fd8fa0bcdeb41d3870c6aea0ac41765 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3363d074de944b6b886e0ad68d0be635 : SUCCESS in 7m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1779e7ff46a94e699e8b52e66d86f76e : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5d3c38100b6a0314b06c8b8a446122682f7c5616","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 15:16:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c0e479ac32c3ffd9fd16affa7375b0147a77a532","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 15:18:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a41793c630e4f8ea7c9cafff4a9befc : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aef32ef24688efe9decff6da9ee19c130bc2a9b7","revisions":{"aef32ef24688efe9decff6da9ee19c130bc2a9b7":{"kind":"REWORK","_number":1,"created":"2020-06-04 09:32:36.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/08/733508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/08/733508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/08/733508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/08/733508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/08/733508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/08/733508/1"}}},"commit":{"parents":[{"commit":"ea0aa97f212ae70714104c3a0923902fb6536cd8","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/ea0aa97f212ae70714104c3a0923902fb6536cd8"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-04 09:31:20.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-04 09:32:18.000000000","tz":120},"subject":"Always reset CONF when starting the wsgi app","message":"Always reset CONF when starting the wsgi app\n\nThis ensures that options loaded during any prior run of the\napplication are dropped before being added again during init_application.\n\nChange-Id: I4267071cdfcb6319ed02e355a9c761fce58e5383\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/aef32ef24688efe9decff6da9ee19c130bc2a9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/aef32ef24688efe9decff6da9ee19c130bc2a9b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
