)]}'
{"id":"openstack%2Fsearchlight~734461","triplet_id":"openstack%2Fsearchlight~master~I130f1a45151dd45426b99869f40b54361278b5c7","project":"openstack/searchlight","branch":"master","topic":"cleanup-py27-support","hashtags":[],"change_id":"I130f1a45151dd45426b99869f40b54361278b5c7","subject":"Cleanup py27 support","status":"MERGED","created":"2020-06-09 08:54:45.000000000","updated":"2020-06-09 14:06:42.000000000","submitted":"2020-06-09 14:04:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734461-1591711470620-f90b450b","meta_rev_id":"dc4e109ff121eea263b1253b6c3964a5cbcfa2e8","_number":734461,"virtual_id_number":734461,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 14:04:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},{"value":2,"date":"2020-06-09 13:46:29.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"date":"2020-06-09 10:04:47.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-09 10:04:47.000000000","_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},{"value":1,"date":"2020-06-09 13:46:29.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 10:04:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-09 10:04:47.000000000","updated_by":{"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},"reviewer":{"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},"state":"REVIEWER"},{"updated":"2020-06-09 13:46:29.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-06-09 14:04:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"64c84741eaecdea26b2888bff1d486529c7ab5cb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 08:54:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9922eab11a214060ef64670fa8a89973205d2dc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 09:16:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/545360abdf1b44e7887ca5ad2dfea701 : SUCCESS in 7m 08s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/817aa32839ad4abfa44990c82addecc7 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f976ee39504c46a536be787e145b94 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdb9ac076ae7462fa4ed72a370289436 : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d78fd4ea74546eb99d439d1d2fd519d : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c1a6c46dde74c6d8df886aa80144b6f : FAILURE in 7m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cb942306ca84e3c8c4a09e452575d85 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be0cff019f94ce79266face7b1488fe : SUCCESS in 7m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/71ca6edac1b049c0b4a8888d0c184902 : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e6620ae63444196aab8cf969565453f : FAILURE in 18m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c175216d1ad04af7ae63537ccb0f6652 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6d4c7a4aa4b61a21334629b4264991a5504fb1a9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 09:24:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3e34aee6d38ff972a83cdaa24a12d86b76f9711b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 09:42:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0bfe1e72812b43739f42d63a700e5097 : SUCCESS in 5m 16s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5077893e56b647f898259e9dd6d6546e : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff4d96ba8df54b6b825ee4f0f1bdd81b : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c09471efa9444c21af9d725e12076c93 : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94301839bc2340f1b46e61045cf2eb02 : SUCCESS in 9m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26480bed154342e297e1dc34a4bb564a : FAILURE in 10m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/823ad0ba65c0441eb2308afd9911d3bd : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21b1500947c4e229d83d83a06460f03 : SUCCESS in 9m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/46c07d367e9745a793266f70f14c951d : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93a4c486b66f45e8b0d71de249e12f1f : FAILURE in 16m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6932b93df5041cc8286ed928565f2bf : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"da407cd5995e7435622209df4daee5f2148dad26","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 09:46:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98f1695acdaac615507f8dae5da21d6718003056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 10:03:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ddd75dbcf6ba43108e32ae7ef9116b0b : SUCCESS in 6m 43s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8bcc2005a7be4009964806deb2ef6368 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a94f2f86933c43d2950c2ae7ff2467db : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5476a66f34aa49f1bbba60c37d021f87 : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87492afea93240f89a31bfeb0bf401fd : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/631d797b20d04b1f990cd5c274e718be : FAILURE in 10m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2658e2e6c4a54f9380e3a240985fdd1f : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81de0c7feb1c49dfb77e012824c3012a : SUCCESS in 12m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c785d9ddd86341dca9dbaa436d5ef782 : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e18b7cfb2b3407daac9b6b88e50b9eb : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/842faf7cdcc344aa9f215cac2a5c7e76 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"c308d2cf9d49ae7864fe979ff86f6ad31e9e3616","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-06-09 13:46:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1dc724aa6ac1ab703bdaa88e544d3232db77d830","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 13:46:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ad3861f7ad5bd5c334dd8ef94afc3b5ab105d07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:04:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2600897a772e38de7b76826a834e0c4046896449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:04:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/885cd89f6fb44517847c2e44cc972c23 : SUCCESS in 5m 20s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e014f7014e8a4f1b933fb9f3c7ae35b8 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4722ac310a442a9a9171df0be44776d : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30021f8820f84ed09ca0bbed3a530b86 : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6da80d1c143b476a9d751b32a8864802 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7ba50a9bc3418ab43c99d6c85ecb80 : SUCCESS in 16m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5c01dcd02fbc44a49141e24ba2e93c8d : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb4c718706fd41faa54292d0a759ad0b : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bf0e2cb84cb40139719cf49c3c15b82 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b0ba74bc41337b700a64b61e0468aad25e8d8bb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:06:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/23ec2d701aaf4c1189627c3332d132d0 : SUCCESS in 1m 29s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bd3bf6506e74f86b1295d3eb058bc42 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3def9e17ae44b4091b1ec0dbe087020 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"17d7de0dda7575cbb10cd37260d5e0070b002ef9","revisions":{"3c175ce7bf7158e893b66988d81fc2781931158b":{"kind":"REWORK","_number":1,"created":"2020-06-09 08:54:45.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/61/734461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/61/734461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/61/734461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/61/734461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/61/734461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/61/734461/1"}}},"commit":{"parents":[{"commit":"aef32ef24688efe9decff6da9ee19c130bc2a9b7","subject":"Always reset CONF when starting the wsgi app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/aef32ef24688efe9decff6da9ee19c130bc2a9b7"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 08:48:18.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 08:53:52.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n- Remove install_command from tox.ini, the default is fine\n- Remove Babel from requirements, it\u0027s not needed for running,\n  remove babel.cfg.\n- remove obsolete file openstack-common.conf.\n- Use TOX_CONSTRAINTS_FILE instead of obsolete UPPER_CONSTRAINTS_FILE.\n- Update hacking to current version\n\nChange-Id: I130f1a45151dd45426b99869f40b54361278b5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/3c175ce7bf7158e893b66988d81fc2781931158b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/3c175ce7bf7158e893b66988d81fc2781931158b"}]},"branch":"refs/heads/master"},"17d7de0dda7575cbb10cd37260d5e0070b002ef9":{"kind":"REWORK","_number":2,"created":"2020-06-09 09:46:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/61/734461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/61/734461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/61/734461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/61/734461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/61/734461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/61/734461/2"}}},"commit":{"parents":[{"commit":"aef32ef24688efe9decff6da9ee19c130bc2a9b7","subject":"Always reset CONF when starting the wsgi app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/aef32ef24688efe9decff6da9ee19c130bc2a9b7"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 08:48:18.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 09:46:13.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n- Remove Babel from requirements, it\u0027s not needed for running,\n  remove babel.cfg.\n- remove obsolete file openstack-common.conf.\n- Use TOX_CONSTRAINTS_FILE instead of obsolete UPPER_CONSTRAINTS_FILE.\n- Update hacking to current version\n\nChange-Id: I130f1a45151dd45426b99869f40b54361278b5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/17d7de0dda7575cbb10cd37260d5e0070b002ef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/17d7de0dda7575cbb10cd37260d5e0070b002ef9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
