)]}'
{"id":"openstack%2Fsearchlight~734484","triplet_id":"openstack%2Fsearchlight~master~I31fd71c8eca9d5be029b778a5ac78e52ab7be2a7","project":"openstack/searchlight","branch":"master","topic":"py38","hashtags":[],"change_id":"I31fd71c8eca9d5be029b778a5ac78e52ab7be2a7","subject":"Fix py38 problem","status":"MERGED","created":"2020-06-09 09:35:18.000000000","updated":"2020-06-09 14:03:43.000000000","submitted":"2020-06-09 14:01:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734484-1591711309857-00e32a87","meta_rev_id":"7aa8f16b5e4c46a61d5714e773f36329e709f307","_number":734484,"virtual_id_number":734484,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 14:01:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},{"value":2,"date":"2020-06-09 13:45:46.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"date":"2020-06-09 09:54:57.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2020-06-09 09:54:57.000000000","_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},{"value":1,"date":"2020-06-09 13:45:46.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 09:54:57.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-09 09:54:57.000000000","updated_by":{"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},"reviewer":{"_account_id":29228,"name":"Thuy Dang","email":"thuydang.de@gmail.com","username":"thuydang"},"state":"REVIEWER"},{"updated":"2020-06-09 13:45:46.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-06-09 14:01:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b4a4a3e54c68386c32a3ee498ff0b51652a0ddd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 09:35:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ac2ad734f20ba0353cf9982c72288aa49012698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 09:49:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b78745ca8554ec19666f178d62c97b8 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85992f94b8f547d78ef90121b0b8933f : SUCCESS in 9m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e2a9f97e28141ddae386f5063810790 : SUCCESS in 12m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/030e578d2db140bf952c59595dcfb9ae : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46a2a57e8e4f4e658f18a237af7b709b : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1f30453afe44c5b02b3a7de949066f : SUCCESS in 11m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/751325e7b777495d96942f60b4067f5c : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d410d55884fcbc5a6dfffb5f90aa672e4c0a5dcd","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-06-09 13:45:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you, Andreas.","accounts_in_message":[],"_revision_number":1},{"id":"8ac167731544979213c716762e133bade9773e90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 13:45:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffac57c9fe0695531e84c908f73efec8f6c988c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:01:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"226fc0d9674b3780fb29b176874c4156085fb6a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:01:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e45b7cb0397448c680a8ab5f93f8f4d0 : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9e1a63890284fba837aba7258190d80 : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66be115550f443c2b55ae9c382bd5ff6 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e9355be66994b7f800d95ddb99ccdf6 : SUCCESS in 14m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/399f74c732344c1ba5e6d426b0776762 : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":1},{"id":"0f3dd0c0e8d2203df44991b818127b23c93b29eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:03:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad9bc91aeaf4a148d1f071cd4b1a0d8 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14798f56b053f8a8819b6f51e26839cc746a3d16","revisions":{"14798f56b053f8a8819b6f51e26839cc746a3d16":{"kind":"REWORK","_number":1,"created":"2020-06-09 09:35:18.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/84/734484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/84/734484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/84/734484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/84/734484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/84/734484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/84/734484/1"}}},"commit":{"parents":[{"commit":"aef32ef24688efe9decff6da9ee19c130bc2a9b7","subject":"Always reset CONF when starting the wsgi app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/aef32ef24688efe9decff6da9ee19c130bc2a9b7"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 09:34:47.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 09:34:47.000000000","tz":120},"subject":"Fix py38 problem","message":"Fix py38 problem\n\npy38 complains about:\nRuntimeError: dictionary keys changed during iteration\n\nFix it.\n\nChange-Id: I31fd71c8eca9d5be029b778a5ac78e52ab7be2a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/14798f56b053f8a8819b6f51e26839cc746a3d16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/14798f56b053f8a8819b6f51e26839cc746a3d16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
