)]}'
{"id":"openstack%2Fsearchlight~750124","triplet_id":"openstack%2Fsearchlight~master~If701a694dff4189256e28afac061f4abc8fb8fbf","project":"openstack/searchlight","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"If701a694dff4189256e28afac061f4abc8fb8fbf","subject":"Fix gate to add babel in requirement","status":"ABANDONED","created":"2020-09-07 06:17:12.000000000","updated":"2020-09-09 09:20:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"86874459242cc7225502167267342504086a7097","_number":750124,"virtual_id_number":750124,"owner":{"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-07 08:07:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2020-09-07 13:55:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":-1,"date":"2020-09-07 21:44:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"date":"2020-09-07 13:19:08.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-07 08:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-07 13:19:08.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-09-07 13:55:57.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-09-07 21:44:11.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"47a408131f725f8df5ddd68140c8a8d05db17171","author":{"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},"date":"2020-09-07 06:17:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89eff4af109b78ad8b1863bde778d2c036bad337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 06:36:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5acfa42926c64fc0a6c993032f67e031 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddef9b1861054432b86a7902f5a40042 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25061a856e304bc8ba7704ece9c7cf6a : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29c2390fb7d945f78f20f0857f1e416e : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3cd03fb5094aed8f5241ffa576b854 : SUCCESS in 8m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8f8775a6b1004370a913b873efbfa810 : FAILURE in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22b68f2737fe47ff8d3a68a381c11126 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"07e168b947a1fffabfd7e0964991a4de135c84c7","author":{"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},"date":"2020-09-07 07:56:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f739099be066e826628f653f163df1dd5904a1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 08:07:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf1787f4c0c4b67841c11ade30a1159 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51ca91682f3f4fdfadc4ae7830a4538d : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3597873d2c6146759c9534f900f208af : SUCCESS in 5m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68965f350e204640967e1648104b29ba : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00538727a67f469090cd829cd7e04007 : SUCCESS in 5m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/df1cac6e732c498aa7515ec47ca1a211 : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f070e787adfd4a9482d2ad805c8feea6 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":2},{"id":"1cd566a4c19c557e7d0f590bdb3df51cff36e951","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-07 13:18:30.000000000","message":"Topic set to migrate-to-focal","accounts_in_message":[],"_revision_number":2},{"id":"de8447cf21aee3ab92e95036c1c93bb6835778b2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-07 13:18:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"419be2b4e88f6f96cc2b8aab445b1d25fdc92817","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-07 13:55:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f3d71684e5cd7b033b2f72cdf71b27fc0f13005b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-07 21:44:11.000000000","message":"Patch Set 2: Code-Review-1\n\njust realised, there is no translation things in searchlight anymore, we should not need babel.","accounts_in_message":[],"_revision_number":2},{"id":"86874459242cc7225502167267342504086a7097","author":{"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},"date":"2020-09-09 09:20:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c944298374164ae1e211a5d389b834dc4e258bb4","revisions":{"2172cfb03cb3312b7b6f6cb1bde9916e2854b127":{"kind":"REWORK","_number":1,"created":"2020-09-07 06:17:12.000000000","uploader":{"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},"ref":"refs/changes/24/750124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/24/750124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/24/750124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/24/750124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/24/750124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/24/750124/1"}}},"commit":{"parents":[{"commit":"2d5d6dd491fba14b43742961fbdad5116d4510ff","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/2d5d6dd491fba14b43742961fbdad5116d4510ff"}]}],"author":{"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","date":"2020-09-07 06:16:52.000000000","tz":330},"committer":{"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","date":"2020-09-07 06:16:52.000000000","tz":330},"subject":"Fix gate to add babel in requirement","message":"Fix gate to add babel in requirement\n\nCurrent gate is failing with below error\n-   ModuleNotFoundError: No module named \u0027babel\u0027\n\nChange-Id: If701a694dff4189256e28afac061f4abc8fb8fbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/2172cfb03cb3312b7b6f6cb1bde9916e2854b127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/2172cfb03cb3312b7b6f6cb1bde9916e2854b127"}]},"branch":"refs/heads/master"},"c944298374164ae1e211a5d389b834dc4e258bb4":{"kind":"REWORK","_number":2,"created":"2020-09-07 07:56:02.000000000","uploader":{"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},"ref":"refs/changes/24/750124/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/24/750124/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/24/750124/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/24/750124/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/24/750124/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/24/750124/2"}}},"commit":{"parents":[{"commit":"2d5d6dd491fba14b43742961fbdad5116d4510ff","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/2d5d6dd491fba14b43742961fbdad5116d4510ff"}]}],"author":{"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","date":"2020-09-07 06:16:52.000000000","tz":330},"committer":{"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","date":"2020-09-07 07:55:46.000000000","tz":330},"subject":"Fix gate to add babel in requirement","message":"Fix gate to add babel in requirement\n\nCurrent gate is failing with below error\n-   ModuleNotFoundError: No module named \u0027babel\u0027\n\nChange-Id: If701a694dff4189256e28afac061f4abc8fb8fbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/c944298374164ae1e211a5d389b834dc4e258bb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/c944298374164ae1e211a5d389b834dc4e258bb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
