)]}'
{"id":"openstack%2Fsearchlight~750218","triplet_id":"openstack%2Fsearchlight~master~I2517de86556268ecfaf2bff751191e4a2aea2abb","project":"openstack/searchlight","branch":"master","hashtags":[],"change_id":"I2517de86556268ecfaf2bff751191e4a2aea2abb","subject":"DNM: testing gate","status":"ABANDONED","created":"2020-09-07 21:45:29.000000000","updated":"2020-12-07 18:30:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2a74fa170059a741e5c8bc80cc2ae58013eb2872","_number":750218,"virtual_id_number":750218,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-07 21:52:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-07 21:52:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c516f93758d2efb23b6fb7e783b9107defbfb0e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-07 21:45:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95bb8b77bf78dbe6fa682a1af4f3d5a65d7d8428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 21:52:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee45793fbd34bd0898dff8d974d1932 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cdd2b1e839e43c08cc71938f14846b3 : FAILURE in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7060be9389247a885a631158cfc21e0 : FAILURE in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b0cd0baf4764ecda8581ba96005c3ce : FAILURE in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b793eafec081470aaafcaeb012ce5fbd : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d521b184066749b787020071455a9dff : FAILURE in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2a74fa170059a741e5c8bc80cc2ae58013eb2872","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-07 18:30:25.000000000","message":"Abandoned\n\nProject Retirement: http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018637.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4ef1488acc5ddc1473335853580252374a9dc65","revisions":{"e4ef1488acc5ddc1473335853580252374a9dc65":{"kind":"REWORK","_number":1,"created":"2020-09-07 21:45:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/18/750218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/18/750218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/18/750218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/18/750218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/18/750218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/18/750218/1"}}},"commit":{"parents":[{"commit":"2d5d6dd491fba14b43742961fbdad5116d4510ff","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/2d5d6dd491fba14b43742961fbdad5116d4510ff"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-07 21:45:10.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-07 21:45:27.000000000","tz":-300},"subject":"DNM: testing gate","message":"DNM: testing gate\n\nChange-Id: I2517de86556268ecfaf2bff751191e4a2aea2abb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/e4ef1488acc5ddc1473335853580252374a9dc65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/e4ef1488acc5ddc1473335853580252374a9dc65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
