)]}'
{"id":"openstack%2Fsearchlight~756317","triplet_id":"openstack%2Fsearchlight~master~I0e93b59cf5cd29d8f3f9a880cff7500534620556","project":"openstack/searchlight","branch":"master","hashtags":[],"change_id":"I0e93b59cf5cd29d8f3f9a880cff7500534620556","subject":"Dep\u0027s should be restricted by upper-constraints","status":"ABANDONED","created":"2020-10-06 14:22:43.000000000","updated":"2020-12-07 18:29:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"24c83762f9eb73ff9960bdbff8928cc9447137c4","_number":756317,"virtual_id_number":756317,"owner":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-06 15:07:39.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-06 15:05:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-06 14:29:59.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 14:29:59.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-10-06 15:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-06 15:07:39.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"}],"messages":[{"id":"03d61c98d29f3bcb344cbbf14f6eab3b31cd46db","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-10-06 14:22:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ffa603698918eb9624ae02600f1b3a1bc66ae2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 15:05:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0349cfbc1cb842438dd225c6beba518b : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c580ee3b9a744eb68b93723014f31d96 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd4e0608b8064df4b76fa4c736765818 : FAILURE in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0dca91e413e443288e9bc20bc1367f9 : FAILURE in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eeaded6995f44e279c3ddce516aacebc : FAILURE in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aacef9df81344498e058eb890203452 : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba72eed735244434b2429625ee4f0b0c : FAILURE in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1405b9241c1c41a085146c43a70c5a98 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"dfc3cd14e674696a9d4b53df329c6899828e4f9c","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-10-06 15:07:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be6b673eb2988edf8ccf2346c351d3c4350b2a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 15:23:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f03c83eac1304cabbbd340924eca4078 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e4bfc092cd4e9587f10a6b89601d5d : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a0c9c4274994a53b38c7b44803c0c7c : FAILURE in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cc22697b23546bfb7a38f7a309fec97 : FAILURE in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c25db25f7e449d3a02bff02be6c9025 : FAILURE in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/239b9a5a06054897b4744c4e16de84f2 : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e528237b4749417ead453bb228a57f8e : FAILURE in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad69a06a9ff641f1a92f525fedee82b0 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"24c83762f9eb73ff9960bdbff8928cc9447137c4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-07 18:29:29.000000000","message":"Abandoned\n\nProject Retirement: http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018637.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49f0bb64473e52c4eb9d113045ce4101175b0b46","revisions":{"49f0bb64473e52c4eb9d113045ce4101175b0b46":{"kind":"REWORK","_number":1,"created":"2020-10-06 14:22:43.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/17/756317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/17/756317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/17/756317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/17/756317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/17/756317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/17/756317/1"}}},"commit":{"parents":[{"commit":"ae76091a3b1ab651f1fbc4f1db8a34a953c99db8","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/ae76091a3b1ab651f1fbc4f1db8a34a953c99db8"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-10-06 14:22:22.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-10-06 14:22:41.000000000","tz":480},"subject":"Dep\u0027s should be restricted by upper-constraints","message":"Dep\u0027s should be restricted by upper-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s upper-constraints.txt\n\nChange-Id: I0e93b59cf5cd29d8f3f9a880cff7500534620556\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/49f0bb64473e52c4eb9d113045ce4101175b0b46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/49f0bb64473e52c4eb9d113045ce4101175b0b46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
