)]}'
{"id":"openstack%2Fsearchlight~764526","triplet_id":"openstack%2Fsearchlight~master~I0c5c9a961cca19113f4d59e37cb16e3bc0ccf8e1","project":"openstack/searchlight","branch":"master","topic":"retire-searchlight","hashtags":[],"change_id":"I0c5c9a961cca19113f4d59e37cb16e3bc0ccf8e1","subject":"Retire the Searchlight project","status":"MERGED","created":"2020-11-28 03:49:12.000000000","updated":"2020-12-07 18:35:18.000000000","submitted":"2020-12-07 18:35:18.000000000","submitter":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764526-retire-searchlight","meta_rev_id":"84ce0193b85a943a38653669d313d4af5f8c87af","_number":764526,"virtual_id_number":764526,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-03 17:33:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-07 18:34:57.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-04 18:25:10.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2020-12-06 11:18:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-07 18:34:57.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-28 03:57:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-06 11:18:03.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"}],"messages":[{"id":"706dcb1cda8114ce4c4fe2541a4e291a12ddb013","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-28 03:49:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8920a08d726e0385fea586c5e50dee1c979e3004","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 03:57:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/79d74f1d15ca424fbe3f99b3ca589b4f : FAILURE in 2m 28s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/72ba411aaa4c4c1cbc8e224311fccaf0 : FAILURE in 1m 42s","accounts_in_message":[],"_revision_number":1},{"id":"98cb04dddcc83fc9b6e69285521098bdac6cfbb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-28 04:02:13.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"59cc91e616297a1c4559b2a08ed785a8019aba11","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-28 04:02:26.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"70f431ed9da21be208fb34e848082f6a129126dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 04:09:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fdd6b1d016654a8583adf2f74ca96ee8 : FAILURE in 4m 21s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ff70530ec8e64c01bf60edd6a5d572f3 : FAILURE in 2m 38s","accounts_in_message":[],"_revision_number":2},{"id":"b86cbeeaa7d2bcd2e47246e601b0011537a8198e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-29 00:27:08.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"0ec4efe7c56882b153379065f3c5556562afa1bd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-03 17:32:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fc90359b85d1a469932fe3135dfa8e1337a5354d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:33:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/81cf3a37c0f145d4b62dba84465adcf6 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"d611b1e295a1644022e352e55468c5223e997d5c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-04 18:25:10.000000000","message":"Patch Set 2: Code-Review+2\n\nlgtm, need another TC member to review and +A.","accounts_in_message":[],"_revision_number":2},{"id":"a67f45ebe34d502f15c3eac7ca7dbe6fde9f465f","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-12-06 11:18:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9cd0ef6779a15cddd179ffa3e126c65ba63af935","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-07 18:34:57.000000000","message":"Patch Set 2: Verified+2 Workflow+1\n\nMerging it as Trinh is +1 on this.\n\nProject Retirement: http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018637.html","accounts_in_message":[],"_revision_number":2},{"id":"84ce0193b85a943a38653669d313d4af5f8c87af","tag":"autogenerated:gerrit:merged","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-07 18:35:18.000000000","message":"Change has been successfully merged by Ghanshyam","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"73316422315176b8b97dcc0c7bccf1fc49d7eb0e","revisions":{"e7e48863140e36b7d6721b0516c3b07f6b931151":{"kind":"REWORK","_number":1,"created":"2020-11-28 03:49:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/26/764526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/26/764526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/26/764526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/26/764526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/26/764526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/26/764526/1"}}},"commit":{"parents":[{"commit":"ae76091a3b1ab651f1fbc4f1db8a34a953c99db8","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/ae76091a3b1ab651f1fbc4f1db8a34a953c99db8"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-28 03:46:40.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-28 03:49:11.000000000","tz":-360},"subject":"Retire the Searchlight project","message":"Retire the Searchlight project\n\nAs announced in openstack-discuss ML[1], Searchlight project\nis retiring in Wallaby cycle.\n\nThis commit retires this repository as per process deinfed in\nproject-guide[2]. Anyone would like to maintain it again, please\nrevert back this commit and propose the re-adding of Searchlight\nto governance.\n\nThe community wishes to express our thanks and appreciation to all of\nthose who have contributed to the Searchlight project over the years.\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/764519\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018637.html\n[2] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nChange-Id: I0c5c9a961cca19113f4d59e37cb16e3bc0ccf8e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/e7e48863140e36b7d6721b0516c3b07f6b931151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/e7e48863140e36b7d6721b0516c3b07f6b931151"}]},"branch":"refs/heads/master"},"73316422315176b8b97dcc0c7bccf1fc49d7eb0e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-28 04:02:13.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/26/764526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/26/764526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/26/764526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/26/764526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/26/764526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/26/764526/2"}}},"commit":{"parents":[{"commit":"ae76091a3b1ab651f1fbc4f1db8a34a953c99db8","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/ae76091a3b1ab651f1fbc4f1db8a34a953c99db8"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-28 03:46:40.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-28 04:01:59.000000000","tz":0},"subject":"Retire the Searchlight project","message":"Retire the Searchlight project\n\nAs announced in openstack-discuss ML[1], Searchlight project\nis retiring in Wallaby cycle.\n\nThis commit retires this repository as per process deinfed in\nproject-guide[2]. Anyone would like to maintain it again, please\nrevert back this commit and propose the re-adding of Searchlight\nto governance.\n\nThe community wishes to express our thanks and appreciation to all of\nthose who have contributed to the Searchlight project over the years.\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/764519\nNeeded-By: https://review.opendev.org/c/openstack/governance/+/764530\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018637.html\n[2] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nChange-Id: I0c5c9a961cca19113f4d59e37cb16e3bc0ccf8e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/73316422315176b8b97dcc0c7bccf1fc49d7eb0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/73316422315176b8b97dcc0c7bccf1fc49d7eb0e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
