)]}'
{"id":"openstack%2Fsenlin-tempest-plugin~666572","triplet_id":"openstack%2Fsenlin-tempest-plugin~master~I5db55fffb1fa2ffe47a6317359b2ed905459426c","project":"openstack/senlin-tempest-plugin","branch":"master","hashtags":[],"change_id":"I5db55fffb1fa2ffe47a6317359b2ed905459426c","subject":"Sync Sphinx requirement","status":"ABANDONED","created":"2019-06-20 13:20:26.000000000","updated":"2020-05-06 05:43:58.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7413d3baa34a238c87e2278ea572a05a8b2085e4","_number":666572,"virtual_id_number":666572,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-25 15:30:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"all":[{"value":1,"date":"2019-10-25 14:28:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 08:07:05.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2019-09-09 06:16:29.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-10-10 07:29:30.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-10-25 14:28:36.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-10-25 15:30:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"475c0bfc05de96148e60fb247875f62a4b885496","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-20 13:20:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6da505dafabf74307c6a7456a876f7114324358b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 15:04:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/72/666572/1/check/test-release-openstack/705de4f/ : SUCCESS in 3m 32s\n- requirements-check http://logs.openstack.org/72/666572/1/check/requirements-check/cea7357/ : SUCCESS in 8m 06s\n- openstack-tox-pep8 http://logs.openstack.org/72/666572/1/check/openstack-tox-pep8/5ebdfea/ : SUCCESS in 4m 50s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/72/666572/1/check/senlin-dsvm-tempest-py27-api/05fa382/ : SUCCESS in 41m 25s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/72/666572/1/check/senlin-dsvm-tempest-py27-functional/1b66194/ : SUCCESS in 1h 31m 23s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/72/666572/1/check/senlin-dsvm-tempest-py27-integration/392700a/ : SUCCESS in 41m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ec6a3f3972241f1370270214e023f6c28260b47","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-07-11 08:07:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cb75c4639361a31e8245e3d05a4b83d49827aa2f","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-07-13 16:19:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"41209b744c4a75eb9a1694ed870176ce2e5a2569","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-03 07:56:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c17f63422c2cf612c0e13f94ae1b2e5e4dc2f3d7","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-09 06:16:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"415eb14bc61bd7e36b24b5694fc1e3320977df02","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-10 07:29:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68c1f0b59b6b654dd42093b3cf9505d758d6b999","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-10-25 14:27:51.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"08eb91b0829dadcba3e005f75168a5123c01fe6f","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-10-25 14:28:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1174b45f122096290863c341412af5631a8405c9","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-10-25 14:28:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6c1745dcd629adf67dafdae57a6fa0ae141d7ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 15:30:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/65fbc596d16f457b9ff0cd83ac251f46 : SUCCESS in 2m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9348a5a98e65471cbfe30548ff3cb4cd : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddc4f2528554495f8ff85965c965fd4e : SUCCESS in 3m 16s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/1cf9de3e69194343a114cd9249307d0a : SUCCESS in 58m 33s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/b8daa343c87a4cdd8f7e96f76b0e9947 : SUCCESS in 41m 29s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/a3274d8b4ec5496594bac515b17f27fe : FAILURE in 47m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7413d3baa34a238c87e2278ea572a05a8b2085e4","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-06 05:43:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"18d63079acd7e148218dd3c7f99ce9f224ebc97f","revisions":{"73291bc471e4bf0dbbc8dd91f28607a089c6b503":{"kind":"REWORK","_number":1,"created":"2019-06-20 13:20:26.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/72/666572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin-tempest-plugin","ref":"refs/changes/72/666572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/72/666572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/72/666572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/72/666572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/72/666572/1"}}},"commit":{"parents":[{"commit":"184d46a24666740ef7507356dd6590e33b1884e9","subject":"Merge \"update git.openstack.org to opendev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/184d46a24666740ef7507356dd6590e33b1884e9"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-20 13:20:08.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-20 13:20:24.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Remove unncessary \"\u003d\u003d\"\n\nChange-Id: I5db55fffb1fa2ffe47a6317359b2ed905459426c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/73291bc471e4bf0dbbc8dd91f28607a089c6b503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/73291bc471e4bf0dbbc8dd91f28607a089c6b503"}]},"branch":"refs/heads/master"},"18d63079acd7e148218dd3c7f99ce9f224ebc97f":{"kind":"REWORK","_number":2,"created":"2019-10-25 14:28:29.000000000","uploader":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"ref":"refs/changes/72/666572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin-tempest-plugin","ref":"refs/changes/72/666572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/72/666572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/72/666572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/72/666572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/72/666572/2"}}},"commit":{"parents":[{"commit":"184d46a24666740ef7507356dd6590e33b1884e9","subject":"Merge \"update git.openstack.org to opendev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/184d46a24666740ef7507356dd6590e33b1884e9"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-20 13:20:08.000000000","tz":480},"committer":{"name":"huangliujie","email":"huang.liujie@99cloud.net","date":"2019-10-25 14:28:24.000000000","tz":0},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Remove unncessary \"\u003d\u003d\"\n\nChange-Id: I5db55fffb1fa2ffe47a6317359b2ed905459426c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/18d63079acd7e148218dd3c7f99ce9f224ebc97f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/18d63079acd7e148218dd3c7f99ce9f224ebc97f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
