)]}'
{"id":"openstack%2Fsenlin-tempest-plugin~690423","triplet_id":"openstack%2Fsenlin-tempest-plugin~master~I283467b3dc4a6f32b1c7b81c2e97a52d8c0c9f1e","project":"openstack/senlin-tempest-plugin","branch":"master","topic":"node_taint","hashtags":[],"change_id":"I283467b3dc4a6f32b1c7b81c2e97a52d8c0c9f1e","subject":"Add tests for new tainted field in nodes","status":"MERGED","created":"2019-10-22 21:31:40.000000000","updated":"2019-10-25 18:51:08.000000000","submitted":"2019-10-25 18:51:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690423-1572029468107-89326a6a","meta_rev_id":"1d4be2e7c711c8aed4e6a184a74e11840169fdbf","_number":690423,"virtual_id_number":690423,"owner":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"date":"2019-10-23 18:04:31.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 18:51:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"all":[{"value":1,"date":"2019-10-22 22:27:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":1,"date":"2019-10-25 14:28:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-23 18:17:26.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-25 18:01:47.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 22:27:15.000000000","updated_by":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"reviewer":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"state":"REVIEWER"},{"updated":"2019-10-25 14:28:57.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-10-25 18:01:47.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-10-25 18:51:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76067d84fd1a8d624e41ee954c7b7d062bc04245","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-22 21:31:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"967e5b98dbe1b17d966ebc904015f4735e843281","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-22 21:34:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b40e82de0ec839b80a2777173585d97953b58e66","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-10-22 22:27:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2102f143cebb79ff1986463a9461e67159f11f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 22:34:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83bfe23eb4894ad3abb60839994c7f88 : SUCCESS in 2m 57s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/cda9e2527b3f4c879ea690c5b8487bd4 : SUCCESS in 45m 39s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/fe378f2df7124fdfa31dff21c93674f7 : SUCCESS in 39m 33s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/8b6d315787c54ec1973890b744aec2f6 : FAILURE in 58m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55858ef0869f2141f96470fc0e1b3c29d26dc8a8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-23 18:02:15.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b0cca804b9b8ee6bfccfa90d5aecc3700fd600b4","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-23 18:04:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb541097f20afe25a843e1bb6add5c91ff22d8a0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-23 18:17:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"571d0ec42c81005d1422c61b46042244b2812238","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-24 04:55:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b3dd188283787b6fabed76db6557e3f189ee2895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 05:55:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c13d1851fca4e71bdab01566ad16532 : SUCCESS in 3m 15s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/ee01c46187754c40a8da97409bf1c5c2 : SUCCESS in 52m 25s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/36b1b72984a747d680aa4c74d9d9cc94 : SUCCESS in 48m 17s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/ec4fa05fd5c1473c9583c4882b5bfdd8 : FAILURE in 58m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e5c78460e20c8202e45be8dec1537ddc01c37c2a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-24 05:58:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cc6efd287c8fb1bdb4c7ae3ede2dd43b1e0eb8ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 07:12:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c05fa7cc9dc4dd0ab2825679806502a : SUCCESS in 2m 53s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/d09dbeaf6dcd4586b7b3f4a7ea1f76da : SUCCESS in 41m 03s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/6af68bdbc57340ae9d8a9cad8c4449cf : SUCCESS in 1h 04m 39s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/e7493c3dcc52400e99d9cbce861caf07 : FAILURE in 46m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e8a3a5fadded45c1be657b31bf06434cc4df2b3","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-10-25 14:28:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c67ced14afbc19ba321c845a1b5315fdbb6bae7f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-25 18:01:47.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fbf3e6cb8194d6ccf0f984fadab271e16ebe4d91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 18:01:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"35444897541bc8b94b4d325fe6fff7838a9b13ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 18:51:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ede7e359d95f4bf3b666ec1c06858024 : SUCCESS in 4m 11s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/f650c689cccb429ba3c6e608f16d1b7b : SUCCESS in 47m 00s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/41f6dead5e414b4ab5f7de148a6a1057 : SUCCESS in 43m 55s","accounts_in_message":[],"_revision_number":2},{"id":"a6ddd18e77f3984741ef3b68c82f46d15d0a6d20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 18:51:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"01aa3de05fc511ef905001fc2e612a69f92bf64c","revisions":{"83c9d28b2a2e8fd928216e36af77e9e72391f1f1":{"kind":"REWORK","_number":1,"created":"2019-10-22 21:31:40.000000000","uploader":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"ref":"refs/changes/23/690423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin-tempest-plugin","ref":"refs/changes/23/690423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/23/690423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/23/690423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/23/690423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/23/690423/1"}}},"commit":{"parents":[{"commit":"e8c68ffb745c241edfe345471b206706eb4914cb","subject":"Switch to the new canonical constraints URL on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/e8c68ffb745c241edfe345471b206706eb4914cb"}]}],"author":{"name":"Duc Truong","email":"dtruong@blizzard.com","date":"2019-10-22 20:03:03.000000000","tz":0},"committer":{"name":"Duc Truong","email":"dtruong@blizzard.com","date":"2019-10-22 21:31:31.000000000","tz":0},"subject":"Add tests for new tainted field in nodes","message":"Add tests for new tainted field in nodes\n\n- Added API tests for nodes returning tainted fields in v1.13+\n- Fixed missing and duplicate idempotent ids.\n\nDepends-On: https://review.opendev.org/#/c/689606/\nChange-Id: I283467b3dc4a6f32b1c7b81c2e97a52d8c0c9f1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/83c9d28b2a2e8fd928216e36af77e9e72391f1f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/83c9d28b2a2e8fd928216e36af77e9e72391f1f1"}]},"branch":"refs/heads/master"},"01aa3de05fc511ef905001fc2e612a69f92bf64c":{"kind":"REWORK","_number":2,"created":"2019-10-22 21:34:15.000000000","uploader":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"ref":"refs/changes/23/690423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin-tempest-plugin","ref":"refs/changes/23/690423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/23/690423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/23/690423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/23/690423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin-tempest-plugin refs/changes/23/690423/2"}}},"commit":{"parents":[{"commit":"e8c68ffb745c241edfe345471b206706eb4914cb","subject":"Switch to the new canonical constraints URL on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/e8c68ffb745c241edfe345471b206706eb4914cb"}]}],"author":{"name":"Duc Truong","email":"dtruong@blizzard.com","date":"2019-10-22 20:03:03.000000000","tz":0},"committer":{"name":"Duc Truong","email":"dtruong@blizzard.com","date":"2019-10-22 21:34:08.000000000","tz":0},"subject":"Add tests for new tainted field in nodes","message":"Add tests for new tainted field in nodes\n\n- Added API tests for nodes returning tainted fields in v1.13+\n- Fixed missing and duplicate idempotent ids.\n\nDepends-On: https://review.opendev.org/#/c/689606/\nChange-Id: I283467b3dc4a6f32b1c7b81c2e97a52d8c0c9f1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/01aa3de05fc511ef905001fc2e612a69f92bf64c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin-tempest-plugin/commit/01aa3de05fc511ef905001fc2e612a69f92bf64c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
