)]}'
{"id":"openstack%2Fsenlin~423012","triplet_id":"openstack%2Fsenlin~master~I659cf440523dc12d5769197e5786c8a1d1428a85","project":"openstack/senlin","branch":"master","topic":"health-policy-lb","hashtags":[],"change_id":"I659cf440523dc12d5769197e5786c8a1d1428a85","subject":"Remove LB_STATUS_POLLING from health detection type","status":"MERGED","created":"2017-01-20 03:11:21.000000000","updated":"2017-01-20 05:59:06.000000000","submitted":"2017-01-20 05:59:06.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"960b8e553000b0b070104b69209bbe6981e01a43","_number":423012,"virtual_id_number":423012,"owner":{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-01-20 05:59:06.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-01-20 05:11:45.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":2,"date":"2017-01-20 05:11:55.000000000","_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":1,"date":"2017-01-20 05:11:55.000000000","_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-01-20 05:11:45.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2017-01-20 05:59:06.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"2c85ab8c9e785cdb8ddfbd950c2985e78a47611f","author":{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},"date":"2017-01-20 03:11:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39b1a55df62d9386e8d88e786c68e9e6a539dda0","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2017-01-20 03:47:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"65d24407ca4ffea9d46e8a6c58ff542fd8a574ca","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-20 03:50:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-senlin-docs-ubuntu-xenial http://docs-draft.openstack.org/12/423012/1/check/gate-senlin-docs-ubuntu-xenial/f073c32//doc/build/html/ : SUCCESS in 2m 04s\n- gate-senlin-pep8-ubuntu-xenial http://logs.openstack.org/12/423012/1/check/gate-senlin-pep8-ubuntu-xenial/19987f4/ : FAILURE in 3m 54s\n- gate-senlin-python27-ubuntu-xenial http://logs.openstack.org/12/423012/1/check/gate-senlin-python27-ubuntu-xenial/091f6c2/ : SUCCESS in 3m 08s\n- gate-senlin-python35 http://logs.openstack.org/12/423012/1/check/gate-senlin-python35/81de720/ : SUCCESS in 2m 51s\n- senlin-coverage-ubuntu-xenial http://logs.openstack.org/12/423012/1/check/senlin-coverage-ubuntu-xenial/4d9b793/ : SUCCESS in 3m 53s (non-voting)\n- gate-senlin-dsvm-tempest-api http://logs.openstack.org/12/423012/1/check/gate-senlin-dsvm-tempest-api/7fe2f0a/ : SUCCESS in 37m 23s\n- gate-senlin-dsvm-tempest-functional http://logs.openstack.org/12/423012/1/check/gate-senlin-dsvm-tempest-functional/a35f61b/ : SUCCESS in 21m 31s\n- gate-senlin-dsvm-tempest-integration-nv http://logs.openstack.org/12/423012/1/check/gate-senlin-dsvm-tempest-integration-nv/f18632b/ : FAILURE in 38m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e40b64eda734f905eaf8af797fe3dbbeea941f8b","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2017-01-20 04:39:46.000000000","message":"Patch Set 2: Published edit on patch set 1","accounts_in_message":[],"_revision_number":2},{"id":"90c877d3d0e625d730da49f2f97935d965128431","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-20 05:09:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-senlin-docs-ubuntu-xenial http://docs-draft.openstack.org/12/423012/2/check/gate-senlin-docs-ubuntu-xenial/372c615//doc/build/html/ : SUCCESS in 2m 16s\n- gate-senlin-pep8-ubuntu-xenial http://logs.openstack.org/12/423012/2/check/gate-senlin-pep8-ubuntu-xenial/2c17d68/ : SUCCESS in 4m 39s\n- gate-senlin-python27-ubuntu-xenial http://logs.openstack.org/12/423012/2/check/gate-senlin-python27-ubuntu-xenial/1a5db9e/ : SUCCESS in 3m 29s\n- gate-senlin-python35 http://logs.openstack.org/12/423012/2/check/gate-senlin-python35/a143acb/ : SUCCESS in 2m 51s\n- senlin-coverage-ubuntu-xenial http://logs.openstack.org/12/423012/2/check/senlin-coverage-ubuntu-xenial/38b7c42/ : SUCCESS in 4m 10s (non-voting)\n- gate-senlin-dsvm-tempest-api http://logs.openstack.org/12/423012/2/check/gate-senlin-dsvm-tempest-api/4f1fa1c/ : SUCCESS in 29m 24s\n- gate-senlin-dsvm-tempest-functional http://logs.openstack.org/12/423012/2/check/gate-senlin-dsvm-tempest-functional/2650ad6/ : SUCCESS in 21m 47s\n- gate-senlin-dsvm-tempest-integration-nv http://logs.openstack.org/12/423012/2/check/gate-senlin-dsvm-tempest-integration-nv/6982caf/ : FAILURE in 22m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"63a44b754d194740999205a89c94973e5f4b691f","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2017-01-20 05:11:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b2c9a925322820fd0a91dfd0e2ead91ef2b6f95e","author":{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},"date":"2017-01-20 05:11:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d487369fd24dd5a800ef72e4eee068d1e3c202f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-20 05:12:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"168f2b957a3ad0bef463c014ab27b5cb0a7f864d","date":"2017-01-20 05:59:06.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"a073b51b266a76f3ae60b07a57ee272293244e95","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-20 05:59:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-senlin-docs-ubuntu-xenial http://docs-draft.openstack.org/12/423012/2/gate/gate-senlin-docs-ubuntu-xenial/643ea4e//doc/build/html/ : SUCCESS in 3m 29s\n- gate-senlin-pep8-ubuntu-xenial http://logs.openstack.org/12/423012/2/gate/gate-senlin-pep8-ubuntu-xenial/a90ba43/ : SUCCESS in 2m 57s\n- gate-senlin-python27-ubuntu-xenial http://logs.openstack.org/12/423012/2/gate/gate-senlin-python27-ubuntu-xenial/0cc7b76/ : SUCCESS in 5m 53s\n- gate-senlin-python35 http://logs.openstack.org/12/423012/2/gate/gate-senlin-python35/f3fb3c1/ : SUCCESS in 5m 10s\n- gate-senlin-dsvm-tempest-api http://logs.openstack.org/12/423012/2/gate/gate-senlin-dsvm-tempest-api/14b5c18/ : SUCCESS in 28m 04s\n- gate-senlin-dsvm-tempest-functional http://logs.openstack.org/12/423012/2/gate/gate-senlin-dsvm-tempest-functional/b0efcbf/ : SUCCESS in 46m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d3147c7d15202dfcfc94fe60b7d6d01c8bba83fd","revisions":{"080920702ca760124daa60f2034f73b4887dec2d":{"kind":"REWORK","_number":1,"created":"2017-01-20 03:11:21.000000000","uploader":{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},"ref":"refs/changes/12/423012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/12/423012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/12/423012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/12/423012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/12/423012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/12/423012/1"}}},"commit":{"parents":[{"commit":"771539b6612312532b03373ec291214939c78cdd","subject":"Merge \"Fix various problems in doc tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/771539b6612312532b03373ec291214939c78cdd"}]}],"author":{"name":"tengqm","email":"tengqim@cn.ibm.com","date":"2017-01-20 03:10:09.000000000","tz":-300},"committer":{"name":"tengqm","email":"tengqim@cn.ibm.com","date":"2017-01-20 03:10:09.000000000","tz":-300},"subject":"Remove LB_STATUS_POLLING from health detection type","message":"Remove LB_STATUS_POLLING from health detection type\n\nSince we don\u0027t have this feature yet due to LBaaS / Ocativia side bugs,\nwe should mask them out in case users want to try it.\n\nChange-Id: I659cf440523dc12d5769197e5786c8a1d1428a85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/080920702ca760124daa60f2034f73b4887dec2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/080920702ca760124daa60f2034f73b4887dec2d"}]},"branch":"refs/heads/master"},"d3147c7d15202dfcfc94fe60b7d6d01c8bba83fd":{"kind":"REWORK","_number":2,"created":"2017-01-20 04:39:46.000000000","uploader":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"ref":"refs/changes/12/423012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/12/423012/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/12/423012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/12/423012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/12/423012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/12/423012/2"}}},"commit":{"parents":[{"commit":"771539b6612312532b03373ec291214939c78cdd","subject":"Merge \"Fix various problems in doc tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/771539b6612312532b03373ec291214939c78cdd"}]}],"author":{"name":"tengqm","email":"tengqim@cn.ibm.com","date":"2017-01-20 03:10:09.000000000","tz":-300},"committer":{"name":"XueFeng Liu","email":"liu.xuefeng1@zte.com.cn","date":"2017-01-20 04:39:43.000000000","tz":0},"subject":"Remove LB_STATUS_POLLING from health detection type","message":"Remove LB_STATUS_POLLING from health detection type\n\nSince we don\u0027t have this feature yet due to LBaaS / Ocativia side bugs,\nwe should mask them out in case users want to try it.\n\nChange-Id: I659cf440523dc12d5769197e5786c8a1d1428a85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d3147c7d15202dfcfc94fe60b7d6d01c8bba83fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d3147c7d15202dfcfc94fe60b7d6d01c8bba83fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
