)]}'
{"id":"openstack%2Fsenlin~657188","triplet_id":"openstack%2Fsenlin~master~I9865e7393f9e539cdb87b6c9ddfa483f5405287b","project":"openstack/senlin","branch":"master","topic":"bug/1826545","hashtags":[],"change_id":"I9865e7393f9e539cdb87b6c9ddfa483f5405287b","subject":"Bypass project ID restriction in LBaaS driver.","status":"MERGED","created":"2019-05-05 02:53:39.000000000","updated":"2019-05-28 17:17:30.000000000","submitted":"2019-05-28 17:17:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657188-1559063850274-8e637ae5","meta_rev_id":"3cd85409969c015cf7da421ed030709562fb4e76","_number":657188,"virtual_id_number":657188,"owner":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"value":2,"date":"2019-05-28 17:17:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"recommended":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":1,"date":"2019-05-23 02:37:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":1,"date":"2019-05-24 00:49:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":2,"date":"2019-05-28 16:27:21.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":2,"date":"2019-05-27 23:11:26.000000000","_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"all":[{"value":0,"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":1,"date":"2019-05-28 16:27:21.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 03:53:33.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-05-23 02:37:54.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"state":"REVIEWER"},{"updated":"2019-05-24 00:49:52.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2019-05-27 23:11:26.000000000","updated_by":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"reviewer":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"state":"REVIEWER"},{"updated":"2019-05-28 16:27:21.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2019-05-28 17:17:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3b0fd46513f6ef8027d9373c00a188c26f103d5","author":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"date":"2019-05-05 02:53:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a82d2e2fd97bf82e78cdb5c3c102855ccc42055","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 03:46:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/88/657188/1/check/openstack-tox-lower-constraints/33638ab/ : SUCCESS in 6m 04s\n- openstack-tox-pep8 http://logs.openstack.org/88/657188/1/check/openstack-tox-pep8/c33e70c/ : SUCCESS in 6m 53s\n- openstack-tox-py27 http://logs.openstack.org/88/657188/1/check/openstack-tox-py27/df117bd/ : SUCCESS in 5m 48s\n- openstack-tox-py36 http://logs.openstack.org/88/657188/1/check/openstack-tox-py36/2c9a0e4/ : SUCCESS in 5m 21s\n- openstack-tox-py37 http://logs.openstack.org/88/657188/1/check/openstack-tox-py37/09aca7c/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/88/657188/1/check/openstack-tox-docs/fd8faaf/html/ : SUCCESS in 4m 51s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/88/657188/1/check/senlin-dsvm-tempest-py27-api/86dc4ee/ : SUCCESS in 46m 21s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/88/657188/1/check/senlin-dsvm-tempest-py3-api/ce3e05a/ : SUCCESS in 47m 25s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/88/657188/1/check/senlin-dsvm-tempest-py27-functional/cf196f9/ : SUCCESS in 38m 22s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/88/657188/1/check/senlin-dsvm-tempest-py3-functional/5c11d0a/ : SUCCESS in 34m 42s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/88/657188/1/check/senlin-dsvm-tempest-py27-integration/d7bd0ad/ : SUCCESS in 45m 07s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/88/657188/1/check/senlin-dsvm-tempest-py3-integration/dacdeac/ : SUCCESS in 40m 29s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/88/657188/1/check/openstack-tox-cover/7c0c59c/cover/ : SUCCESS in 7m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c1805e7a97d2c31241d17c4fa9960e0a8ef34be","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2019-05-06 08:25:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0d1b9318b5257a6375957fd652d62c77910a403e","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-05-06 21:56:10.000000000","message":"Patch Set 1:\n\nI\u0027m trying to understand the problem here.  Are you saying that the context returned by oslo_context.get_current() does not contain a project_id field?","accounts_in_message":[],"_revision_number":1},{"id":"6aededc22d73902391b8cf26749d07170580ea56","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2019-05-07 00:47:08.000000000","message":"Patch Set 1:\n\n\u003e I\u0027m trying to understand the problem here.  Are you saying that the\n \u003e context returned by oslo_context.get_current() does not contain a\n \u003e project_id field?\n\nYou can see this bug: https://bugs.launchpad.net/senlin/+bug/1826545.","accounts_in_message":[],"_revision_number":1},{"id":"664aa14bba03bae3813b3ebdc799457aaaa882d7","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-05-07 03:53:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1379dc2603d7862c149ebdac39bcc02ec0ea6794","author":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"date":"2019-05-13 19:15:01.000000000","message":"Patch Set 1:\n\n\u003e I\u0027m trying to understand the problem here.  Are you saying that the\n \u003e context returned by oslo_context.get_current() does not contain a\n \u003e project_id field?\n\nYes, that is correct.","accounts_in_message":[],"_revision_number":1},{"id":"f748ec9af1258f44158dd23ba548ff59bd0a6ec3","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-05-22 13:50:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18217856c14e4ef4fc6ae4d28348e6e1ed3222cf","author":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"date":"2019-05-22 17:18:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"674b09d5f8b5d60ba69ca9e353c8c8bc06064c32","author":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"date":"2019-05-22 17:18:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84e391390b96fb5409343f1d7aff6bc3cd1b2bcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 20:11:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/88/657188/2/check/openstack-tox-lower-constraints/8f99333/ : SUCCESS in 7m 24s\n- openstack-tox-pep8 http://logs.openstack.org/88/657188/2/check/openstack-tox-pep8/0d042fd/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/88/657188/2/check/openstack-tox-py27/b866045/ : SUCCESS in 6m 00s\n- openstack-tox-py36 http://logs.openstack.org/88/657188/2/check/openstack-tox-py36/ae1f35a/ : SUCCESS in 6m 27s\n- openstack-tox-py37 http://logs.openstack.org/88/657188/2/check/openstack-tox-py37/2782213/ : SUCCESS in 6m 17s\n- openstack-tox-docs http://logs.openstack.org/88/657188/2/check/openstack-tox-docs/3c3e181/html/ : SUCCESS in 5m 16s\n- senlin-dsvm-tempest-py27-api finger://ze03.openstack.org/1194fd64eb4a4f84b0ad9fd6b2830aef : RETRY_LIMIT in 43m 37s\n- senlin-dsvm-tempest-py3-api finger://ze04.openstack.org/67678db44b714b2aae05f05df4bf02cc : RETRY_LIMIT in 32m 38s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/88/657188/2/check/senlin-dsvm-tempest-py27-functional/a0df013/ : POST_FAILURE in 42m 18s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/88/657188/2/check/senlin-dsvm-tempest-py3-functional/0210042/ : POST_FAILURE in 42m 20s\n- senlin-dsvm-tempest-py27-integration finger://ze03.openstack.org/47f3918e926b48f8b64439af00dbe4c7 : RETRY_LIMIT in 32m 57s (non-voting)\n- senlin-dsvm-tempest-py3-integration finger://ze05.openstack.org/431a98ef257e4576b848ef3e4ad1be66 : RETRY_LIMIT in 40m 32s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/88/657188/2/check/openstack-tox-cover/b7c3864/cover/ : SUCCESS in 8m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5226c35c62cc56fc71f6ea42d50b950a1eaed04","author":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"date":"2019-05-22 21:57:46.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"61670ee4cbe3a5b4d2f4ad9e0a67d88b2ac9d05a","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-05-22 22:26:49.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)\n\nI tested it and verified that you fix works.  We can get rid of the node_obj and directly use node variable.  I tested that and it works ok for me.","accounts_in_message":[],"_revision_number":3},{"id":"de31eecabb8908c2576984b5644592be376c14e6","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-05-22 22:27:19.000000000","message":"Patch Set 3:\n\n\u003e (3 comments)\n \u003e \n \u003e I tested it and verified that you fix works.  We can get rid of the\n \u003e node_obj and directly use node variable.  I tested that and it\n \u003e works ok for me.\n\ns/you/your","accounts_in_message":[],"_revision_number":3},{"id":"48773afc2959edc507a5f28f9f113deb8b47c6d8","author":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"date":"2019-05-22 22:54:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1d422114fb7cb3ef21e9146ed0e30a70f8e5fee2","author":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"date":"2019-05-22 22:54:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c0561e99b1605de0cb1eae93f01f2e7a8971c861","author":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"date":"2019-05-22 23:10:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1d9a3e00d01f3be1d49a131156d9d16abb465c85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 00:18:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/88/657188/5/check/openstack-tox-lower-constraints/1a2798b/ : SUCCESS in 6m 05s\n- openstack-tox-pep8 http://logs.openstack.org/88/657188/5/check/openstack-tox-pep8/46969f5/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/88/657188/5/check/openstack-tox-py27/78c856c/ : SUCCESS in 7m 58s\n- openstack-tox-py36 http://logs.openstack.org/88/657188/5/check/openstack-tox-py36/51ea30d/ : SUCCESS in 5m 43s\n- openstack-tox-py37 http://logs.openstack.org/88/657188/5/check/openstack-tox-py37/f6fcb07/ : SUCCESS in 8m 08s\n- openstack-tox-docs http://logs.openstack.org/88/657188/5/check/openstack-tox-docs/67d03ea/html/ : SUCCESS in 6m 16s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/88/657188/5/check/senlin-dsvm-tempest-py27-api/d7770f1/ : SUCCESS in 46m 57s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/88/657188/5/check/senlin-dsvm-tempest-py3-api/dae953e/ : SUCCESS in 43m 12s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/88/657188/5/check/senlin-dsvm-tempest-py27-functional/00e7a8e/ : SUCCESS in 38m 02s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/88/657188/5/check/senlin-dsvm-tempest-py3-functional/4581cfa/ : SUCCESS in 39m 21s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/88/657188/5/check/senlin-dsvm-tempest-py27-integration/725ca60/ : SUCCESS in 40m 57s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/88/657188/5/check/senlin-dsvm-tempest-py3-integration/91c4c95/ : SUCCESS in 47m 30s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/88/657188/5/check/openstack-tox-cover/9b93e46/cover/ : SUCCESS in 8m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b9f5c704ae84a576887594553c65dd3daf0ccd17","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2019-05-23 02:37:54.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d16a57a954f872ef2af449af354b85cfc1dfe617","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-05-23 15:53:48.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d7ac399114996f2880a717b47b78708a0afb8a9a","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-05-23 16:07:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f9e2ff2c46864419b57338fc4d9b6d25849bde85","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-05-24 00:49:52.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"67954cee2265ef668f07ec792d4fb4dbfe73388c","author":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"date":"2019-05-27 23:11:26.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"490c0fe52455ae01e4605707d11e690e40061e73","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-05-28 16:27:21.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c5eb8c1b97262722268af7327c5a45fe1e506ff1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 16:27:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"20f77ec6eec102d111f0c3eed8b9a7e5686cdcc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 17:17:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/88/657188/5/gate/openstack-tox-lower-constraints/1cbee3e/ : SUCCESS in 5m 54s\n- openstack-tox-pep8 http://logs.openstack.org/88/657188/5/gate/openstack-tox-pep8/2b5537f/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/88/657188/5/gate/openstack-tox-py27/aeaadfa/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/88/657188/5/gate/openstack-tox-py36/f28ec7b/ : SUCCESS in 5m 42s\n- openstack-tox-py37 http://logs.openstack.org/88/657188/5/gate/openstack-tox-py37/b99996c/ : SUCCESS in 6m 56s\n- openstack-tox-docs http://logs.openstack.org/88/657188/5/gate/openstack-tox-docs/49b31a4/html/ : SUCCESS in 4m 30s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/88/657188/5/gate/senlin-dsvm-tempest-py27-api/9b9d3bb/ : SUCCESS in 42m 47s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/88/657188/5/gate/senlin-dsvm-tempest-py27-functional/064d4b8/ : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":5},{"id":"8619ed6ff02b9e153e515e7a2b9ffa297e89f6d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 17:17:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"83234c0c47b5aa6678fd4cb175b4e24e1e158cd6","revisions":{"4e6ff47d4e737913c9c9af0c262debeae6da8204":{"kind":"REWORK","_number":1,"created":"2019-05-05 02:53:39.000000000","uploader":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"ref":"refs/changes/88/657188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/88/657188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/88/657188/1"}}},"commit":{"parents":[{"commit":"4764595ac7163cd6630d14ac1ff6612bd5ae7ffd","subject":"Allow updates to be issued to degraded LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/4764595ac7163cd6630d14ac1ff6612bd5ae7ffd"}]}],"author":{"name":"Jacob Estelle","email":"jacobestelle@oath.com","date":"2019-05-05 02:45:22.000000000","tz":-420},"committer":{"name":"Jacob Estelle","email":"jacobestelle@oath.com","date":"2019-05-05 02:53:37.000000000","tz":-420},"subject":"Bypass project ID restriction in LBaaS driver","message":"Bypass project ID restriction in LBaaS driver\n\nCalling get_details on a node object requires querying for profile based\non project ID. However the context returned from get_current does not\nhave a project ID set so fails the lookup. Bypass by calling with\nproject_safe\u003dFalse.\n\nChange-Id: I9865e7393f9e539cdb87b6c9ddfa483f5405287b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/4e6ff47d4e737913c9c9af0c262debeae6da8204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/4e6ff47d4e737913c9c9af0c262debeae6da8204"}]},"branch":"refs/heads/master"},"c88fc82cfbce33016d511abf17361ca05e44be4e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-22 17:18:03.000000000","uploader":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"ref":"refs/changes/88/657188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/88/657188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/88/657188/2"}}},"commit":{"parents":[{"commit":"4764595ac7163cd6630d14ac1ff6612bd5ae7ffd","subject":"Allow updates to be issued to degraded LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/4764595ac7163cd6630d14ac1ff6612bd5ae7ffd"}]}],"author":{"name":"Jacob Estelle","email":"jacobestelle@oath.com","date":"2019-05-05 02:45:22.000000000","tz":-420},"committer":{"name":"Jacob Estelle","email":"jacobestelle@verizonmedia.com","date":"2019-05-22 17:17:50.000000000","tz":-420},"subject":"Bypass project ID restriction in LBaaS driver","message":"Bypass project ID restriction in LBaaS driver\n\nCalling get_details on a node object requires querying for profile based\non project ID. However the context returned from get_current does not\nhave a project ID set so fails the lookup. Bypass by calling with\nproject_safe\u003dFalse.\n\nCloses-Bug: #1826545\n\nChange-Id: I9865e7393f9e539cdb87b6c9ddfa483f5405287b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c88fc82cfbce33016d511abf17361ca05e44be4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c88fc82cfbce33016d511abf17361ca05e44be4e"}]},"branch":"refs/heads/master"},"c25dc3d407269ab1734dcc9beb4c5765681ea485":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-22 21:57:46.000000000","uploader":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"ref":"refs/changes/88/657188/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/88/657188/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/88/657188/3"}}},"commit":{"parents":[{"commit":"4764595ac7163cd6630d14ac1ff6612bd5ae7ffd","subject":"Allow updates to be issued to degraded LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/4764595ac7163cd6630d14ac1ff6612bd5ae7ffd"}]}],"author":{"name":"Jacob Estelle","email":"jacobestelle@oath.com","date":"2019-05-05 02:45:22.000000000","tz":-420},"committer":{"name":"Jacob Estelle","email":"jacobestelle@verizonmedia.com","date":"2019-05-22 21:57:21.000000000","tz":-420},"subject":"Bypass project ID restriction in LBaaS driver.","message":"Bypass project ID restriction in LBaaS driver.\n\nCalling get_details on a node object requires querying for profile based\non project ID. However the context returned from get_current does not\nhave a project ID set so fails the lookup. Bypass by calling with\nproject_safe\u003dFalse.\n\nCloses-Bug: #1826545\n\nChange-Id: I9865e7393f9e539cdb87b6c9ddfa483f5405287b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c25dc3d407269ab1734dcc9beb4c5765681ea485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c25dc3d407269ab1734dcc9beb4c5765681ea485"}]},"branch":"refs/heads/master"},"20857433af66063362037892b98036bae9f46bbd":{"kind":"REWORK","_number":4,"created":"2019-05-22 22:54:33.000000000","uploader":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"ref":"refs/changes/88/657188/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/88/657188/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/88/657188/4"}}},"commit":{"parents":[{"commit":"4764595ac7163cd6630d14ac1ff6612bd5ae7ffd","subject":"Allow updates to be issued to degraded LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/4764595ac7163cd6630d14ac1ff6612bd5ae7ffd"}]}],"author":{"name":"Jacob Estelle","email":"jacobestelle@oath.com","date":"2019-05-05 02:45:22.000000000","tz":-420},"committer":{"name":"Jacob Estelle","email":"jacobestelle@verizonmedia.com","date":"2019-05-22 22:54:23.000000000","tz":-420},"subject":"Bypass project ID restriction in LBaaS driver.","message":"Bypass project ID restriction in LBaaS driver.\n\nCalling get_details on a node object requires querying for profile based\non project ID. However the context returned from get_current does not\nhave a project ID set so fails the lookup. Bypass by calling with\nproject_safe\u003dFalse.\n\nCloses-Bug: #1826545\n\nChange-Id: I9865e7393f9e539cdb87b6c9ddfa483f5405287b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/20857433af66063362037892b98036bae9f46bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/20857433af66063362037892b98036bae9f46bbd"}]},"branch":"refs/heads/master"},"83234c0c47b5aa6678fd4cb175b4e24e1e158cd6":{"kind":"REWORK","_number":5,"created":"2019-05-22 23:10:49.000000000","uploader":{"_account_id":25138,"name":"pheartheceal","email":"pheartheceal@gmail.com","username":"pheartheceal"},"ref":"refs/changes/88/657188/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/88/657188/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/88/657188/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/88/657188/5"}}},"commit":{"parents":[{"commit":"4764595ac7163cd6630d14ac1ff6612bd5ae7ffd","subject":"Allow updates to be issued to degraded LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/4764595ac7163cd6630d14ac1ff6612bd5ae7ffd"}]}],"author":{"name":"Jacob Estelle","email":"jacobestelle@oath.com","date":"2019-05-05 02:45:22.000000000","tz":-420},"committer":{"name":"Jacob Estelle","email":"jacobestelle@verizonmedia.com","date":"2019-05-22 23:10:41.000000000","tz":-420},"subject":"Bypass project ID restriction in LBaaS driver.","message":"Bypass project ID restriction in LBaaS driver.\n\nCalling get_details on a node object requires querying for profile based\non project ID. However the context returned from get_current does not\nhave a project ID set so fails the lookup. Bypass by calling with\nproject_safe\u003dFalse.\n\nCloses-Bug: #1826545\n\nChange-Id: I9865e7393f9e539cdb87b6c9ddfa483f5405287b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/83234c0c47b5aa6678fd4cb175b4e24e1e158cd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/83234c0c47b5aa6678fd4cb175b4e24e1e158cd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
