)]}'
{"id":"openstack%2Fsenlin~657564","triplet_id":"openstack%2Fsenlin~master~Iedc9b65eef35c0f6a16cb7caa34269a9a63c2a87","project":"openstack/senlin","branch":"master","topic":"bug/1830152","hashtags":[],"change_id":"Iedc9b65eef35c0f6a16cb7caa34269a9a63c2a87","subject":"Ignore LB not existed when delete member pool","status":"MERGED","created":"2019-05-07 10:28:36.000000000","updated":"2019-09-13 04:00:44.000000000","submitted":"2019-09-13 03:59:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657564-1568347169630-e46e57b4","meta_rev_id":"c2a51669368925530a30be4b78c284634fc4220e","_number":657564,"virtual_id_number":657564,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"date":"2019-09-13 02:06:25.000000000","_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-09-13 03:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"recommended":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"all":[{"value":1,"date":"2019-06-20 07:16:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2019-09-11 17:03:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-09-01 14:59:55.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":1,"date":"2019-07-14 15:24:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":1,"date":"2019-09-12 05:17:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":0,"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-09-12 05:34:24.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 10:28:47.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-05-22 22:57:57.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2019-06-20 07:16:56.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"state":"REVIEWER"},{"updated":"2019-07-14 15:24:10.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-09-11 17:03:36.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-09-12 05:17:27.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-09-12 05:34:24.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2019-09-13 03:59:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7caa7498d85fd30999522a818e34bf5d0e908dfc","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-05-07 10:28:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be01f4ffc94ccdfc5dbd47de4de04be09984c876","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-05-07 10:28:47.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"f935c4f5aad22289c84b65fe86702b93cf43e64c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 11:22:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/64/657564/1/check/openstack-tox-lower-constraints/b7a7c04/ : SUCCESS in 9m 37s\n- openstack-tox-pep8 http://logs.openstack.org/64/657564/1/check/openstack-tox-pep8/d1080ea/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/64/657564/1/check/openstack-tox-py27/8d5805b/ : SUCCESS in 6m 56s\n- openstack-tox-py36 http://logs.openstack.org/64/657564/1/check/openstack-tox-py36/2c460bb/ : SUCCESS in 6m 04s\n- openstack-tox-py37 http://logs.openstack.org/64/657564/1/check/openstack-tox-py37/656f367/ : SUCCESS in 7m 27s\n- openstack-tox-docs http://logs.openstack.org/64/657564/1/check/openstack-tox-docs/8a09e81/html/ : SUCCESS in 5m 25s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/64/657564/1/check/senlin-dsvm-tempest-py27-api/f0c302b/ : SUCCESS in 45m 43s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/64/657564/1/check/senlin-dsvm-tempest-py3-api/199be49/ : SUCCESS in 47m 18s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/64/657564/1/check/senlin-dsvm-tempest-py27-functional/8745fe1/ : SUCCESS in 51m 27s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/64/657564/1/check/senlin-dsvm-tempest-py3-functional/d8bdd51/ : SUCCESS in 52m 55s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/64/657564/1/check/senlin-dsvm-tempest-py27-integration/dabe3e1/ : SUCCESS in 42m 58s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/64/657564/1/check/senlin-dsvm-tempest-py3-integration/37067cf/ : SUCCESS in 32m 31s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/64/657564/1/check/openstack-tox-cover/34619b3/cover/ : SUCCESS in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5ebd2397e9ab8f0d4069b8f4c694b5b5c84d57b","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-05-09 22:28:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94500e34f5820de6869d92711ab2c0e3f5a59552","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-05-10 02:05:06.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e (1 comment)\n\nYes,\ni get an error when i trying detach a LB policy being attached in a cluster before because i was deleted this LB so i can\u0027t detach this policy","accounts_in_message":[],"_revision_number":1},{"id":"3b7975d85600d5f9bf1da10077667afbb191fc86","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-05-22 22:57:57.000000000","message":"Patch Set 1:\n\nCan you create a bug report and describe the steps to reproduce this bug?\n\nIt performed the following steps and cannot reproduce your problem:\n\n1. Create LB policy\n2. Attach LB policy to cluster that has nodes\n3. Make sure nodes IPs are added to LB pool\n4. Manually delete LB\n5. Detach LB policy\n\nI get this message and no errors:\nINFO senlin.engine.event [None req-b33fa342-2263-45cb-ab90-dd621f32b9c2 None\nNone] test1[be216dec] CLUSTER_DETACH_POLICY[6a36f36c] start: 6a36f36c\nDEBUG senlin.drivers.os.lbaas [None req-b33fa342-2263-45cb-ab90-dd621f32b9c2\nNone None] Loadbalancer (205dc14f-590e-444f-bbc1-d178fc3c9f59) is not\nexisting. {{(pid\u003d25432) lb_delete\n/opt/stack/senlin/senlin/drivers/os/lbaas.py:209}}\nINFO senlin.engine.event [None req-b33fa342-2263-45cb-ab90-dd621f32b9c2 None\nNone] test1[be216dec] CLUSTER_DETACH_POLICY[6a36f36c] end: Policy detached.","accounts_in_message":[],"_revision_number":1},{"id":"3d6fd3a86363295f7a4d7d58021814c3962a8086","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-05-23 03:55:49.000000000","message":"Patch Set 1:\n\n\u003e Can you create a bug report and describe the steps to reproduce\n \u003e this bug?\n \u003e \n \u003e It performed the following steps and cannot reproduce your problem:\n \u003e \n \u003e 1. Create LB policy\n \u003e 2. Attach LB policy to cluster that has nodes\n \u003e 3. Make sure nodes IPs are added to LB pool\n \u003e 4. Manually delete LB\n \u003e 5. Detach LB policy\n \u003e \n \u003e I get this message and no errors:\n \u003e INFO senlin.engine.event [None req-b33fa342-2263-45cb-ab90-dd621f32b9c2\n \u003e None\n \u003e None] test1[be216dec] CLUSTER_DETACH_POLICY[6a36f36c] start:\n \u003e 6a36f36c\n \u003e DEBUG senlin.drivers.os.lbaas [None req-b33fa342-2263-45cb-ab90-dd621f32b9c2\n \u003e None None] Loadbalancer (205dc14f-590e-444f-bbc1-d178fc3c9f59) is\n \u003e not\n \u003e existing. {{(pid\u003d25432) lb_delete\n \u003e /opt/stack/senlin/senlin/drivers/os/lbaas.py:209}}\n \u003e INFO senlin.engine.event [None req-b33fa342-2263-45cb-ab90-dd621f32b9c2\n \u003e None\n \u003e None] test1[be216dec] CLUSTER_DETACH_POLICY[6a36f36c] end: Policy\n \u003e detached.\n\nThanks you for care my problem. I was created a new bug in here https://bugs.launchpad.net/senlin/+bug/1830152","accounts_in_message":[],"_revision_number":1},{"id":"c9bd6495bb6fbacfba83b7e52ae03458d91414ee","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-05-23 06:08:42.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e \u003e Can you create a bug report and describe the steps to reproduce\n \u003e \u003e this bug?\n \u003e \u003e\n \u003e \u003e It performed the following steps and cannot reproduce your\n \u003e problem:\n \u003e \u003e\n \u003e \u003e 1. Create LB policy\n \u003e \u003e 2. Attach LB policy to cluster that has nodes\n \u003e \u003e 3. Make sure nodes IPs are added to LB pool\n \u003e \u003e 4. Manually delete LB\n \u003e \u003e 5. Detach LB policy\n \u003e \u003e\n \u003e \u003e I get this message and no errors:\n \u003e \u003e INFO senlin.engine.event [None req-b33fa342-2263-45cb-ab90-dd621f32b9c2\n \u003e \u003e None\n \u003e \u003e None] test1[be216dec] CLUSTER_DETACH_POLICY[6a36f36c] start:\n \u003e \u003e 6a36f36c\n \u003e \u003e DEBUG senlin.drivers.os.lbaas [None req-b33fa342-2263-45cb-ab90-dd621f32b9c2\n \u003e \u003e None None] Loadbalancer (205dc14f-590e-444f-bbc1-d178fc3c9f59) is\n \u003e \u003e not\n \u003e \u003e existing. {{(pid\u003d25432) lb_delete\n \u003e \u003e /opt/stack/senlin/senlin/drivers/os/lbaas.py:209}}\n \u003e \u003e INFO senlin.engine.event [None req-b33fa342-2263-45cb-ab90-dd621f32b9c2\n \u003e \u003e None\n \u003e \u003e None] test1[be216dec] CLUSTER_DETACH_POLICY[6a36f36c] end: Policy\n \u003e \u003e detached.\n \u003e \n \u003e Thanks you for care my problem. I was created a new bug in here\n \u003e https://bugs.launchpad.net/senlin/+bug/1830152\n\nYou need add: Closes-Bug: #1830152 in you commit message.","accounts_in_message":[],"_revision_number":1},{"id":"ed5b18ab7dbacb97245ae7a21508bd4623e1163b","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-05-23 07:18:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d994a4262ed1f04661f362a34fc957a17d0bf738","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 08:12:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/64/657564/2/check/openstack-tox-lower-constraints/8b5d598/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/64/657564/2/check/openstack-tox-pep8/0396b2b/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/64/657564/2/check/openstack-tox-py27/22fdc69/ : SUCCESS in 5m 48s\n- openstack-tox-py36 http://logs.openstack.org/64/657564/2/check/openstack-tox-py36/27cb24d/ : SUCCESS in 5m 55s\n- openstack-tox-py37 http://logs.openstack.org/64/657564/2/check/openstack-tox-py37/55362a2/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/64/657564/2/check/openstack-tox-docs/7b528b3/html/ : SUCCESS in 7m 07s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/64/657564/2/check/senlin-dsvm-tempest-py27-api/6224fc7/ : SUCCESS in 40m 49s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/64/657564/2/check/senlin-dsvm-tempest-py3-api/6412436/ : SUCCESS in 41m 34s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/64/657564/2/check/senlin-dsvm-tempest-py27-functional/7bad805/ : SUCCESS in 36m 57s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/64/657564/2/check/senlin-dsvm-tempest-py3-functional/6b07fab/ : SUCCESS in 45m 06s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/64/657564/2/check/senlin-dsvm-tempest-py27-integration/ffcdaf5/ : SUCCESS in 45m 45s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/64/657564/2/check/senlin-dsvm-tempest-py3-integration/85186d3/ : SUCCESS in 40m 06s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/64/657564/2/check/openstack-tox-cover/8cfa649/cover/ : SUCCESS in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"700a5490c8d33c4887e372ca4a0c9ef1a6e8e38f","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2019-06-20 07:16:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c781aed58d4e281b9270095722a89530e9885bb2","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-07-14 15:24:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5d757a8d0be2ea467511144162eb3dc46176e2f2","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-09-01 14:59:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0188137c82d34f274f69f5f2d813c3080bc67ba1","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-09-11 17:03:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a5c4eabcf7d9a4a9cf7220236b0ab24aad446d84","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-09-12 05:17:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"eca342011de43d78189782d4529644122bb5b8b3","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-09-12 05:34:24.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b4655f81184602566614d81d52626061c1a7c0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 05:34:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ea09a5b4b2e98e204d3a47f36b9ba6983d197af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 05:49:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c54c7b977c9b42d7a2293d77c25df589 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515aa122f0664bf6816aa39eb536a2aa : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/944f7f8317144354b3b2e5f3bfa765c7 : FAILURE in 2m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea82721bea034c44a7995d0e118d2185 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c44d67609f104e15b4e37e8c3ad7c1b2 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c72dc8d0cd84bcca50ed1f09e7db93e : SUCCESS in 4m 12s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/9786bd4df22c44c4bd54b8710015034b : FAILURE in 13m 33s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/e377dc1c19a246bc878da822d82cef6b : FAILURE in 14m 15s","accounts_in_message":[],"_revision_number":2},{"id":"9825ab149a7cfbf9824a55c19c64459f82399931","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-09-13 02:06:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9b6a0feb7888fcb01f2def1a6b0b6fed6d720954","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 03:06:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4825f98a5c1a4bb5982cd8192a17f8ee : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15921ee1598f4c64b60a8805f04e38bc : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/94246e2190d54c63a402c6c63e7e31cf : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8480c04aca24e978f511f63da1a555e : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbea2884ffae4cb384163b6950af1172 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e666de1abbf4c3f9dbd8e7dfb47056e : SUCCESS in 5m 22s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/70707012a0344e719188e5803f60df05 : SUCCESS in 44m 18s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/c6908f24bc4e4e3f96678c4d1fb0b9e6 : SUCCESS in 45m 46s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/a465dae995fb49a8a6c6cd423f62ead7 : SUCCESS in 42m 15s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/7d8729c755bd4323816d4c003ae34bc2 : SUCCESS in 39m 14s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/92494804cccf492d8e4cf6852238a7c8 : SUCCESS in 35m 40s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/275265b936de414b8bc358501360772a : SUCCESS in 47m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cb8ca2665d54bce8524815b79692ed3 : SUCCESS in 7m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"061bce5e9b0fb416dfc3d8a04c25ae44a11e85e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 03:06:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"119d124a10d15b2bb5dc32d19435c50c264afde2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 03:59:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/988647330dc7484697c9339437e50f1b : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab046cdaae747908ebb7a7212135e0d : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bdcde68822f42cc90ee36fd364ee66a : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/613957320a064a8a987268743f48e131 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/955dabf353114e72a453816a53d49b38 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6470989988ce42d29718b4d75f38e21b : SUCCESS in 4m 10s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/427addf767ac436cb8f18395f299a437 : SUCCESS in 47m 05s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/633dac89ae634a3e934d2b0714572457 : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":2},{"id":"26a17f86b0fc9df6c4aa9ccee7ad10be1edc5ada","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 03:59:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2ae0fb5555a4be7003d2c99f1c73d3ac8516a072","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 04:00:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58d8d2f49c4249fd8006318b88f62477 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8cc0503203d811cb1b3cb065681b096025d9825c","revisions":{"2ef65c38399245716352ac74697774df4946ae15":{"kind":"REWORK","_number":1,"created":"2019-05-07 10:28:36.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/64/657564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/64/657564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/64/657564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/64/657564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/64/657564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/64/657564/1"}}},"commit":{"parents":[{"commit":"06c160365e529288d9fa12921f8a3e5c2aad48a3","subject":"Merge \"Move node replacement candidates to dict in inputs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/06c160365e529288d9fa12921f8a3e5c2aad48a3"}]}],"author":{"name":"Bo Tran","email":"yeuconay.caconaynua.96.nd@gmail.com","date":"2019-05-07 10:21:27.000000000","tz":420},"committer":{"name":"Bo Tran","email":"yeuconay.caconaynua.96.nd@gmail.com","date":"2019-05-07 10:21:27.000000000","tz":420},"subject":"Ignore LB not existed when delete member pool","message":"Ignore LB not existed when delete member pool\n\nA Cluster can\u0027t detach a LB policy was attached before\nwhen i do deleted LB have id is specifying in that LB policy.\n\nChange-Id: Iedc9b65eef35c0f6a16cb7caa34269a9a63c2a87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/2ef65c38399245716352ac74697774df4946ae15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/2ef65c38399245716352ac74697774df4946ae15"}]},"branch":"refs/heads/master"},"8cc0503203d811cb1b3cb065681b096025d9825c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-23 07:18:17.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/64/657564/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/64/657564/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/64/657564/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/64/657564/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/64/657564/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/64/657564/2"}}},"commit":{"parents":[{"commit":"06c160365e529288d9fa12921f8a3e5c2aad48a3","subject":"Merge \"Move node replacement candidates to dict in inputs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/06c160365e529288d9fa12921f8a3e5c2aad48a3"}]}],"author":{"name":"Bo Tran","email":"yeuconay.caconaynua.96.nd@gmail.com","date":"2019-05-07 10:21:27.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2019-05-23 07:17:15.000000000","tz":420},"subject":"Ignore LB not existed when delete member pool","message":"Ignore LB not existed when delete member pool\n\nA Cluster can\u0027t detach a LB policy was attached before\nwhen i do deleted LB have id is specifying in that LB policy.\n\nCloses-Bug: #1830152\nChange-Id: Iedc9b65eef35c0f6a16cb7caa34269a9a63c2a87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/8cc0503203d811cb1b3cb065681b096025d9825c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/8cc0503203d811cb1b3cb065681b096025d9825c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
