)]}'
{"id":"openstack%2Fsenlin~657701","triplet_id":"openstack%2Fsenlin~master~I9c723516a65a43533e0589bc85bd485a6387711b","project":"openstack/senlin","branch":"master","topic":"delete_with_detach","hashtags":[],"change_id":"I9c723516a65a43533e0589bc85bd485a6387711b","subject":"Allow cluster delete to detach policies \u0026 delete receivers","status":"MERGED","created":"2019-05-07 22:56:07.000000000","updated":"2019-05-17 09:35:32.000000000","submitted":"2019-05-17 09:35:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657701-1558085732746-a5bd43bf","meta_rev_id":"ad8a33c304d153d07b71edef3d7942fa6d794171","_number":657701,"virtual_id_number":657701,"owner":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"date":"2019-05-08 23:42:25.000000000","_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"value":2,"date":"2019-05-17 09:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":2,"date":"2019-05-17 08:46:38.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":2,"date":"2019-05-09 22:25:46.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":1,"date":"2019-05-17 08:46:38.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 22:25:46.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2019-05-17 08:46:38.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2019-05-17 09:35:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"382473f192d2ab214a311f28e02eec2be4184e08","author":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"date":"2019-05-07 22:56:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f9deb2f195cb1af8040cdb5f773fa2a5c49378e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 23:48:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/01/657701/1/check/openstack-tox-lower-constraints/3fe1af5/ : SUCCESS in 6m 05s\n- openstack-tox-pep8 http://logs.openstack.org/01/657701/1/check/openstack-tox-pep8/3527969/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/01/657701/1/check/openstack-tox-py27/84d89dd/ : SUCCESS in 6m 03s\n- openstack-tox-py36 http://logs.openstack.org/01/657701/1/check/openstack-tox-py36/d50900e/ : SUCCESS in 5m 54s\n- openstack-tox-py37 http://logs.openstack.org/01/657701/1/check/openstack-tox-py37/7adbc7e/ : SUCCESS in 8m 35s\n- openstack-tox-docs http://logs.openstack.org/01/657701/1/check/openstack-tox-docs/bba743b/html/ : SUCCESS in 6m 07s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py27-api/abd1acc/ : FAILURE in 42m 16s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py3-api/1168c93/ : FAILURE in 44m 56s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py27-functional/38c4a45/ : SUCCESS in 36m 39s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py3-functional/b8bb928/ : SUCCESS in 40m 25s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py27-integration/5b0d352/ : SUCCESS in 39m 10s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py3-integration/17f6dec/ : SUCCESS in 40m 34s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/01/657701/1/check/openstack-tox-cover/a6edba5/cover/ : SUCCESS in 7m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5fb2b01f75bd9edce55f0c4e8237564e326672c","author":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"date":"2019-05-08 00:31:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b4ebe64e3694641fe7e9b684a4a0129764dc5a8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 01:19:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/01/657701/1/check/openstack-tox-lower-constraints/0a3415a/ : SUCCESS in 5m 11s\n- openstack-tox-pep8 http://logs.openstack.org/01/657701/1/check/openstack-tox-pep8/d1fded1/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/01/657701/1/check/openstack-tox-py27/fd37161/ : SUCCESS in 5m 30s\n- openstack-tox-py36 http://logs.openstack.org/01/657701/1/check/openstack-tox-py36/27ad5c8/ : SUCCESS in 5m 16s\n- openstack-tox-py37 http://logs.openstack.org/01/657701/1/check/openstack-tox-py37/6764acb/ : SUCCESS in 5m 50s\n- openstack-tox-docs http://logs.openstack.org/01/657701/1/check/openstack-tox-docs/b200460/html/ : SUCCESS in 4m 25s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py27-api/248ba54/ : FAILURE in 42m 19s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py3-api/5ec7eb0/ : FAILURE in 41m 44s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py27-functional/9092c52/ : SUCCESS in 36m 40s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py3-functional/8da3527/ : SUCCESS in 36m 44s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py27-integration/c23a88d/ : SUCCESS in 39m 52s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/01/657701/1/check/senlin-dsvm-tempest-py3-integration/3a83978/ : SUCCESS in 40m 35s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/01/657701/1/check/openstack-tox-cover/bb78ee5/ : FAILURE in 7m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17ae82796dcf40ce93db17870648e71acaae0026","author":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"date":"2019-05-08 01:25:26.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d0056f55c945ea9d4378a4207d02628f17191ee3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 02:27:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/01/657701/2/check/openstack-tox-lower-constraints/a992633/ : SUCCESS in 5m 22s\n- openstack-tox-pep8 http://logs.openstack.org/01/657701/2/check/openstack-tox-pep8/b6dac86/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/01/657701/2/check/openstack-tox-py27/1a5dd7f/ : SUCCESS in 7m 08s\n- openstack-tox-py36 http://logs.openstack.org/01/657701/2/check/openstack-tox-py36/a66ee42/ : SUCCESS in 5m 37s\n- openstack-tox-py37 http://logs.openstack.org/01/657701/2/check/openstack-tox-py37/986074b/ : SUCCESS in 7m 27s\n- openstack-tox-docs http://logs.openstack.org/01/657701/2/check/openstack-tox-docs/76feee8/html/ : SUCCESS in 4m 50s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py27-api/008057d/ : SUCCESS in 47m 41s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py3-api/0ea2406/ : SUCCESS in 45m 14s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py27-functional/b5ba24c/ : SUCCESS in 53m 37s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py3-functional/e88bd9c/ : SUCCESS in 44m 05s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py27-integration/0770fd2/ : FAILURE in 1h 00m 46s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py3-integration/2cd5666/ : SUCCESS in 39m 54s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/01/657701/2/check/openstack-tox-cover/240e1d7/cover/ : SUCCESS in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"205180c9acb57e0e8099bef3ffb3cfa931384b83","author":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"date":"2019-05-08 04:02:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0da8efdb2bebfde4db8b546fb946936d0a3c070a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 04:07:48.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"46814a259b331ede740bc7501d873f80309f0874","author":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"date":"2019-05-08 06:03:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f6d195705d351656ac1754ea98f0697337da0fc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 07:20:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/01/657701/2/check/openstack-tox-lower-constraints/fc01b4d/ : SUCCESS in 6m 07s\n- openstack-tox-pep8 http://logs.openstack.org/01/657701/2/check/openstack-tox-pep8/c79560c/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/01/657701/2/check/openstack-tox-py27/5a47869/ : SUCCESS in 5m 58s\n- openstack-tox-py36 http://logs.openstack.org/01/657701/2/check/openstack-tox-py36/a2feead/ : SUCCESS in 6m 07s\n- openstack-tox-py37 http://logs.openstack.org/01/657701/2/check/openstack-tox-py37/5ac6e0c/ : SUCCESS in 6m 25s\n- openstack-tox-docs http://logs.openstack.org/01/657701/2/check/openstack-tox-docs/e93b921/html/ : SUCCESS in 4m 23s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py27-api/a7fd1b2/ : SUCCESS in 43m 41s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py3-api/471354d/ : SUCCESS in 59m 52s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py27-functional/d763311/ : SUCCESS in 40m 56s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py3-functional/c09948a/ : SUCCESS in 41m 44s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py27-integration/02e896b/ : SUCCESS in 41m 24s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/01/657701/2/check/senlin-dsvm-tempest-py3-integration/c9e7c97/ : SUCCESS in 42m 23s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/01/657701/2/check/openstack-tox-cover/b5264a7/ : FAILURE in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"456a9bb8ada85d64c3c10ce8d83699569388810a","author":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"date":"2019-05-08 22:43:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa72e966904eebab4e7cb79ddc62b4a380c51653","author":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"date":"2019-05-08 23:04:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6b6f91542aec3f14f8b5ce9443b29435fc76f1ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 23:30:29.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"bdcb8cdc87c341fadc85f0bc9f3a71ef432c32f5","author":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"date":"2019-05-08 23:42:25.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3115c50a9cf0ef3e6b98643e15f1bb46e62b8707","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 00:31:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/01/657701/4/check/build-openstack-api-ref/87c0d37/html/ : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/657701/4/check/openstack-tox-lower-constraints/8493ccc/ : SUCCESS in 5m 48s\n- openstack-tox-pep8 http://logs.openstack.org/01/657701/4/check/openstack-tox-pep8/febd056/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/01/657701/4/check/openstack-tox-py27/6fae7be/ : SUCCESS in 5m 26s\n- openstack-tox-py36 http://logs.openstack.org/01/657701/4/check/openstack-tox-py36/1bf6494/ : SUCCESS in 5m 36s\n- openstack-tox-py37 http://logs.openstack.org/01/657701/4/check/openstack-tox-py37/30a8e66/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/01/657701/4/check/openstack-tox-docs/ade98c9/html/ : SUCCESS in 4m 18s\n- build-openstack-releasenotes http://logs.openstack.org/01/657701/4/check/build-openstack-releasenotes/8e12487/html/ : SUCCESS in 8m 31s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/01/657701/4/check/senlin-dsvm-tempest-py27-api/99d1b33/ : SUCCESS in 43m 33s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/01/657701/4/check/senlin-dsvm-tempest-py3-api/16e5da5/ : SUCCESS in 41m 18s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/01/657701/4/check/senlin-dsvm-tempest-py27-functional/4b9b759/ : SUCCESS in 37m 48s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/01/657701/4/check/senlin-dsvm-tempest-py3-functional/245a5f9/ : SUCCESS in 39m 55s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/01/657701/4/check/senlin-dsvm-tempest-py27-integration/e4c48b7/ : SUCCESS in 39m 41s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/01/657701/4/check/senlin-dsvm-tempest-py3-integration/dab5fe0/ : SUCCESS in 38m 50s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/01/657701/4/check/openstack-tox-cover/f03b399/ : FAILURE in 7m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d685508aabdb9d00420ad9b198d330d46370432b","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-05-09 22:25:46.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"474f598273f40795dc81c632e79a920a81054c04","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-05-17 08:46:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"014d025127c869d2fa8affcdd23206759b989175","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 08:46:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0726f70f4cc4b7dbf949e93266286397703b8c5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 09:35:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2d92cd14d1ded1ebaba60bcea11588db3963dbbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 09:35:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/01/657701/4/gate/build-openstack-api-ref/d9d56e8/html/ : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/657701/4/gate/openstack-tox-lower-constraints/c5c2fd2/ : SUCCESS in 5m 44s\n- openstack-tox-pep8 http://logs.openstack.org/01/657701/4/gate/openstack-tox-pep8/2784207/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/01/657701/4/gate/openstack-tox-py27/e4f123e/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/01/657701/4/gate/openstack-tox-py36/56e489a/ : SUCCESS in 5m 43s\n- openstack-tox-py37 http://logs.openstack.org/01/657701/4/gate/openstack-tox-py37/5e59fd3/ : SUCCESS in 5m 50s\n- openstack-tox-docs http://logs.openstack.org/01/657701/4/gate/openstack-tox-docs/9fddabf/html/ : SUCCESS in 4m 36s\n- build-openstack-releasenotes http://logs.openstack.org/01/657701/4/gate/build-openstack-releasenotes/c5bb217/html/ : SUCCESS in 8m 41s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/01/657701/4/gate/senlin-dsvm-tempest-py27-api/e1698bf/ : SUCCESS in 43m 21s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/01/657701/4/gate/senlin-dsvm-tempest-py27-functional/e011969/ : SUCCESS in 37m 11s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e37e3477715b8e9c3a600678132621e70b53a51b","revisions":{"e2d7e614d93880ffdefce50da726566a7b980651":{"kind":"REWORK","_number":1,"created":"2019-05-07 22:56:07.000000000","uploader":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"ref":"refs/changes/01/657701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/01/657701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/01/657701/1"}}},"commit":{"parents":[{"commit":"1a3e61aeab01c0aa1bc71478d50538a611455ee7","subject":"Add unit test for node.do_recover","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/1a3e61aeab01c0aa1bc71478d50538a611455ee7"}]}],"author":{"name":"Jude Cross","email":"jucross@blizzard.com","date":"2019-05-07 22:41:10.000000000","tz":-420},"committer":{"name":"Jude Cross","email":"jucross@blizzard.com","date":"2019-05-07 22:41:21.000000000","tz":-420},"subject":"[WIP] Allow cluster delete to detach policies \u0026 delete receivers","message":"[WIP] Allow cluster delete to detach policies \u0026 delete receivers\n\nThis patch allows the cluster delete actions to detach policies\nand delete receivers for the cluster being deleted. This\nsimplifies deleting clusters by not having to detach or delete\nall dependancies from it beforehand.\n\nChange-Id: I9c723516a65a43533e0589bc85bd485a6387711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e2d7e614d93880ffdefce50da726566a7b980651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e2d7e614d93880ffdefce50da726566a7b980651"}]},"branch":"refs/heads/master"},"fdfe08186ed2cd9391a2de77ffcde10745475bf2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-08 01:25:26.000000000","uploader":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"ref":"refs/changes/01/657701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/01/657701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/01/657701/2"}}},"commit":{"parents":[{"commit":"1a3e61aeab01c0aa1bc71478d50538a611455ee7","subject":"Add unit test for node.do_recover","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/1a3e61aeab01c0aa1bc71478d50538a611455ee7"}]}],"author":{"name":"Jude Cross","email":"jucross@blizzard.com","date":"2019-05-07 22:41:10.000000000","tz":-420},"committer":{"name":"Jude Cross","email":"jucross@blizzard.com","date":"2019-05-08 01:19:49.000000000","tz":-420},"subject":"[WIP] Allow cluster delete to detach policies \u0026 delete receivers","message":"[WIP] Allow cluster delete to detach policies \u0026 delete receivers\n\nThis patch allows the cluster delete actions to detach policies\nand delete receivers for the cluster being deleted. This\nsimplifies deleting clusters by not having to detach or delete\nall dependancies from it beforehand.\n\nDepends-On: https://review.opendev.org/657713/\n\nChange-Id: I9c723516a65a43533e0589bc85bd485a6387711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/fdfe08186ed2cd9391a2de77ffcde10745475bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/fdfe08186ed2cd9391a2de77ffcde10745475bf2"}]},"branch":"refs/heads/master"},"ebba36cc376fdce510ffd904bde7d4f7778cc06c":{"kind":"REWORK","_number":3,"created":"2019-05-08 22:43:09.000000000","uploader":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"ref":"refs/changes/01/657701/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/01/657701/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/01/657701/3"}}},"commit":{"parents":[{"commit":"1a3e61aeab01c0aa1bc71478d50538a611455ee7","subject":"Add unit test for node.do_recover","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/1a3e61aeab01c0aa1bc71478d50538a611455ee7"}]}],"author":{"name":"Jude Cross","email":"jucross@blizzard.com","date":"2019-05-07 22:41:10.000000000","tz":-420},"committer":{"name":"Jude Cross","email":"jucross@blizzard.com","date":"2019-05-08 22:41:20.000000000","tz":-420},"subject":"[WIP] Allow cluster delete to detach policies \u0026 delete receivers","message":"[WIP] Allow cluster delete to detach policies \u0026 delete receivers\n\nThis patch allows the cluster delete actions to detach policies\nand delete receivers for the cluster being deleted. This\nsimplifies deleting clusters by not having to detach or delete\nall dependancies from it beforehand.\n\nDepends-On: https://review.opendev.org/657713/\n\nChange-Id: I9c723516a65a43533e0589bc85bd485a6387711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ebba36cc376fdce510ffd904bde7d4f7778cc06c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ebba36cc376fdce510ffd904bde7d4f7778cc06c"}]},"branch":"refs/heads/master"},"e37e3477715b8e9c3a600678132621e70b53a51b":{"kind":"REWORK","_number":4,"created":"2019-05-08 23:04:17.000000000","uploader":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"ref":"refs/changes/01/657701/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/01/657701/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/657701/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/01/657701/4"}}},"commit":{"parents":[{"commit":"1a3e61aeab01c0aa1bc71478d50538a611455ee7","subject":"Add unit test for node.do_recover","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/1a3e61aeab01c0aa1bc71478d50538a611455ee7"}]}],"author":{"name":"Jude Cross","email":"jucross@blizzard.com","date":"2019-05-07 22:41:10.000000000","tz":-420},"committer":{"name":"Jude Cross","email":"jucross@blizzard.com","date":"2019-05-08 23:03:06.000000000","tz":-420},"subject":"Allow cluster delete to detach policies \u0026 delete receivers","message":"Allow cluster delete to detach policies \u0026 delete receivers\n\nThis patch allows the cluster delete actions to detach policies\nand delete receivers for the cluster being deleted. This\nsimplifies deleting clusters by not having to detach or delete\nall dependancies from it beforehand.\n\nDepends-On: https://review.opendev.org/657713/\n\nChange-Id: I9c723516a65a43533e0589bc85bd485a6387711b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e37e3477715b8e9c3a600678132621e70b53a51b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e37e3477715b8e9c3a600678132621e70b53a51b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
