)]}'
{"id":"openstack%2Fsenlin~659924","triplet_id":"openstack%2Fsenlin~master~Id014484697ab932ea01c47f8054055f3c626be79","project":"openstack/senlin","branch":"master","topic":"add_admin_permission","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-05-15 03:21:01.000000000","reason":"Change was abandoned"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-05-15 03:21:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id014484697ab932ea01c47f8054055f3c626be79","subject":"Add support a user admin can do everything","status":"ABANDONED","created":"2019-05-18 18:44:52.000000000","updated":"2024-05-15 03:21:01.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0a7a8716e2b001463591de6622643650f5088da7","_number":659924,"virtual_id_number":659924,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-23 03:13:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-18 19:50:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-19 20:13:25.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"af5626c5697f22c40d3cc664786063e87068f832","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-05-18 18:44:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96b5baac0c10d1fa2c96dc051865a5a04f645c5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 19:50:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/24/659924/1/check/openstack-tox-lower-constraints/baec842/ : SUCCESS in 8m 24s\n- openstack-tox-pep8 http://logs.openstack.org/24/659924/1/check/openstack-tox-pep8/d0eac7b/ : SUCCESS in 8m 10s\n- openstack-tox-py27 http://logs.openstack.org/24/659924/1/check/openstack-tox-py27/aad28e1/ : SUCCESS in 7m 13s\n- openstack-tox-py36 http://logs.openstack.org/24/659924/1/check/openstack-tox-py36/ec116cf/ : SUCCESS in 6m 27s\n- openstack-tox-py37 http://logs.openstack.org/24/659924/1/check/openstack-tox-py37/6276c79/ : SUCCESS in 6m 31s\n- openstack-tox-docs http://logs.openstack.org/24/659924/1/check/openstack-tox-docs/290ccf6/html/ : SUCCESS in 4m 22s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/24/659924/1/check/senlin-dsvm-tempest-py27-api/834d12f/ : FAILURE in 41m 16s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/24/659924/1/check/senlin-dsvm-tempest-py3-api/9b34b06/ : FAILURE in 38m 46s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/24/659924/1/check/senlin-dsvm-tempest-py27-functional/789fa28/ : FAILURE in 35m 48s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/24/659924/1/check/senlin-dsvm-tempest-py3-functional/80c5e4b/ : FAILURE in 53m 50s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/24/659924/1/check/senlin-dsvm-tempest-py27-integration/801ce9a/ : FAILURE in 37m 26s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/24/659924/1/check/senlin-dsvm-tempest-py3-integration/9e96be5/ : FAILURE in 36m 20s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/24/659924/1/check/openstack-tox-cover/efa0a84/cover/ : SUCCESS in 9m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79a2f9c075ccf3f7f150b7048efe08de643e17bf","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-05-20 07:39:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"b719cc5fc031d61f5dcef4f02457f2e1b865f789","tag":"autogenerated:gerrit:restore","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-01-10 02:38:05.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"207d1952bbf2697128f88f7a060e3ca401d1a737","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-19 20:13:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4650144620bf0104c94f0206e5b2224b608d5f71","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-01-20 01:36:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b6fcf56d1b3abd31c79a325516afd56178b8dad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-01-22 09:44:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6c2b7a3af79d48a92fd85842f065519323c64b4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 11:00:54.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2917d3328d1c40ab9fde736603828f08\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c65f9c5bdfb4aae9db6771c6c445a33 : FAILURE in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0aed8e511ef14ef3a776ec572df7895c : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0eb3f4c1857d473b83b84bbb866fc6d9 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c48e33b3e24473979cecce6fee4719 : SUCCESS in 4m 06s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/2a23b826689a49b1991dc29bc04c1622 : FAILURE in 43m 34s\n- senlin-dsvm-tempest-py3-api-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d53bd7b2c5774b5a996ec29d6a338074 : FAILURE in 55m 35s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/e66a7fee18274d529e1f6b65584f118b : FAILURE in 44m 28s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/c1b67b4085864dc8b93c54d8baa62fa2 : FAILURE in 51m 09s\n- senlin-dsvm-tempest-py3-functional-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9b514594a169452fb4dee01475c4fd10 : FAILURE in 35m 02s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/e8c9820decbf4077ba875f079d5b902e : FAILURE in 1h 01m 10s\n- senlin-dsvm-tempest-py3-integration-zaqar https://zuul.opendev.org/t/openstack/build/55fb71c2b0a5433d977942623018f868 : FAILURE in 41m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c52299b728845dc94588320b5e2293f : SUCCESS in 5m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca5d4a4859a936a2baf3ac8ac2101c3800180903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-01-23 01:53:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b0633df2fd88c2546c7611eb4daad72cee79b8a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 03:13:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d711ca9f95d4573ad87bd7e0ebf6a3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9b1be02ae744fa9e8a5001db8b685d : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9990c74216544cedaa2ae438e13b41e9 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fb64b83d97e4de2800d4f015297a822 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1c167674014e7ca1f29c2abf9cf534 : SUCCESS in 4m 19s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/2e7b529120f84804a3f141cf502af2e2 : FAILURE in 52m 51s\n- senlin-dsvm-tempest-py3-api-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/05412d2564df48db8b6d5c52b1d0d3cb : FAILURE in 1h 09m 41s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/cb75872624a84463931298f70ce057bc : FAILURE in 58m 03s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/1bd304d14fd5447ab6f9b1464a60c14d : FAILURE in 55m 50s\n- senlin-dsvm-tempest-py3-functional-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f40c32a9f61e412fbc99b602c5d65cd3 : FAILURE in 43m 33s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/4fddb0d048a841e1bba5041145d76fb4 : FAILURE in 51m 53s\n- senlin-dsvm-tempest-py3-integration-zaqar https://zuul.opendev.org/t/openstack/build/c32d28331c9543c593ade35bb2ad707f : FAILURE in 38m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7f5786e9b814901b912fccdb25ed54e : SUCCESS in 6m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0a7a8716e2b001463591de6622643650f5088da7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-15 03:21:01.000000000","message":"Abandoned\n\nproject has been retired now.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"54edddbf5846c21f4d0100fc0df6b5f94213465d","revisions":{"9559ee84e01fe990b9a413e1094d128b0670d8d0":{"kind":"REWORK","_number":1,"created":"2019-05-18 18:44:52.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/24/659924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/24/659924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/24/659924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/24/659924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/24/659924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/24/659924/1"}}},"commit":{"parents":[{"commit":"c41f0bfaf649d2cf764c056c2844cd6a5f957cfd","subject":"Add support a user admin can see details of profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c41f0bfaf649d2cf764c056c2844cd6a5f957cfd"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2019-05-18 18:42:07.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2019-05-18 18:44:18.000000000","tz":420},"subject":"Add support a user admin can do everything","message":"Add support a user admin can do everything\n\nA user with role admin can delete, update cluster, policy, ...\nand attach, detach a policy to a cluster and see policy biding\nin a cluster\n\nChange-Id: Id014484697ab932ea01c47f8054055f3c626be79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/9559ee84e01fe990b9a413e1094d128b0670d8d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/9559ee84e01fe990b9a413e1094d128b0670d8d0"}]},"branch":"refs/heads/master"},"52426b2793c347f1290a9dd13da6d0bbc3787c36":{"kind":"REWORK","_number":2,"created":"2024-01-22 09:44:11.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/24/659924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/24/659924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/24/659924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/24/659924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/24/659924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/24/659924/2"}}},"commit":{"parents":[{"commit":"7fa6ae780cb7e9c72a2df1ef0c11d543cd1ffca0","subject":"Merge \"Removing CLUSTER_RESIZE from pre-op skip List at min threshold\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7fa6ae780cb7e9c72a2df1ef0c11d543cd1ffca0"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2019-05-18 18:42:07.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-01-22 09:42:41.000000000","tz":420},"subject":"Add support a user admin can do everything","message":"Add support a user admin can do everything\n\nA user with role admin can delete, update cluster, policy, ...\nand attach, detach a policy to a cluster and see policy biding\nin a cluster\n\nChange-Id: Id014484697ab932ea01c47f8054055f3c626be79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/52426b2793c347f1290a9dd13da6d0bbc3787c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/52426b2793c347f1290a9dd13da6d0bbc3787c36"}]},"branch":"refs/heads/master"},"54edddbf5846c21f4d0100fc0df6b5f94213465d":{"kind":"REWORK","_number":3,"created":"2024-01-23 01:53:06.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/24/659924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/24/659924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/24/659924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/24/659924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/24/659924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/24/659924/3"}}},"commit":{"parents":[{"commit":"7fa6ae780cb7e9c72a2df1ef0c11d543cd1ffca0","subject":"Merge \"Removing CLUSTER_RESIZE from pre-op skip List at min threshold\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7fa6ae780cb7e9c72a2df1ef0c11d543cd1ffca0"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2019-05-18 18:42:07.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-01-23 01:52:55.000000000","tz":420},"subject":"Add support a user admin can do everything","message":"Add support a user admin can do everything\n\nA user with role admin can delete, update cluster, policy, ...\nand attach, detach a policy to a cluster and see policy biding\nin a cluster\n\nChange-Id: Id014484697ab932ea01c47f8054055f3c626be79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/54edddbf5846c21f4d0100fc0df6b5f94213465d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/54edddbf5846c21f4d0100fc0df6b5f94213465d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
