)]}'
{"id":"openstack%2Fsenlin~663556","triplet_id":"openstack%2Fsenlin~master~I8bb297c4497b83e06d558a812219d2a56c2e1ee1","project":"openstack/senlin","branch":"master","hashtags":[],"change_id":"I8bb297c4497b83e06d558a812219d2a56c2e1ee1","subject":"Cap sphinx for py2 to match global requirements","status":"ABANDONED","created":"2019-06-06 09:21:10.000000000","updated":"2019-08-06 20:42:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7628fda65773160d47af6f165bcb77c6140f0a99","_number":663556,"virtual_id_number":663556,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2019-07-14 15:32:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":1,"date":"2019-07-14 15:23:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":1,"date":"2019-07-19 00:40:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-14 15:23:51.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-07-14 15:32:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-19 00:40:35.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"265185899c5cc4c8479c73d1b18b742e4a4a2cee","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-06-06 09:21:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f251ba814eb3d2d8d1a031a13aa7674f8760a6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:34:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/56/663556/1/check/build-openstack-api-ref/4c414ef/ : POST_FAILURE in 3m 59s\n- requirements-check http://logs.openstack.org/56/663556/1/check/requirements-check/c991a75/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/56/663556/1/check/openstack-tox-pep8/c934d64/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/56/663556/1/check/openstack-tox-docs/903069b/html/ : SUCCESS in 4m 45s\n- build-openstack-releasenotes http://logs.openstack.org/56/663556/1/check/build-openstack-releasenotes/992e809/html/ : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"25a81ddd964dcdb968febbb234705ef347a9efd5","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-07-14 15:23:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"af91ad4dfc9f0a9563329a01251582019073e51f","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-07-14 15:23:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9dd2f4bf49e6d3c9ab4476ba3a61866bc295a9a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 15:32:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/56/663556/2/check/build-openstack-api-ref/be52b67/html/ : SUCCESS in 3m 01s\n- requirements-check http://logs.openstack.org/56/663556/2/check/requirements-check/bb7b08d/ : SUCCESS in 2m 56s\n- openstack-tox-pep8 http://logs.openstack.org/56/663556/2/check/openstack-tox-pep8/69bb159/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/56/663556/2/check/openstack-tox-docs/35c60c0/html/ : SUCCESS in 3m 28s\n- build-openstack-releasenotes http://logs.openstack.org/56/663556/2/check/build-openstack-releasenotes/0d150a6/html/ : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"753fca107c5ecba163ed3f7aa248e040d59cdefe","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-07-19 00:40:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7628fda65773160d47af6f165bcb77c6140f0a99","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-08-06 20:42:02.000000000","message":"Abandoned\n\nChange was already made in https://review.opendev.org/#/c/673966/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c331dda25f5b1536702d0ecc237097bcb23ad376","revisions":{"05a7b5c38aebe8c9b054a4c0bb5f2e72ebff2216":{"kind":"REWORK","_number":1,"created":"2019-06-06 09:21:10.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/56/663556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/56/663556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/56/663556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/56/663556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/56/663556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/56/663556/1"}}},"commit":{"parents":[{"commit":"80ba25ec6ac1c35787802163f6e42e7625422618","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/80ba25ec6ac1c35787802163f6e42e7625422618"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-06 09:20:39.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-06 09:21:05.000000000","tz":480},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: I8bb297c4497b83e06d558a812219d2a56c2e1ee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/05a7b5c38aebe8c9b054a4c0bb5f2e72ebff2216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/05a7b5c38aebe8c9b054a4c0bb5f2e72ebff2216"}]},"branch":"refs/heads/master"},"c331dda25f5b1536702d0ecc237097bcb23ad376":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-14 15:23:43.000000000","uploader":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"ref":"refs/changes/56/663556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/56/663556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/56/663556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/56/663556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/56/663556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/56/663556/2"}}},"commit":{"parents":[{"commit":"5da4c6830f4f3fbca3f73e92d326a7b0f1e1f720","subject":"Merge \"Switch to the new canonical constraints URL on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/5da4c6830f4f3fbca3f73e92d326a7b0f1e1f720"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-06 09:20:39.000000000","tz":480},"committer":{"name":"yanghongyu","email":"yang.hongyu@99cloud.net","date":"2019-07-14 15:23:43.000000000","tz":0},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: I8bb297c4497b83e06d558a812219d2a56c2e1ee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c331dda25f5b1536702d0ecc237097bcb23ad376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c331dda25f5b1536702d0ecc237097bcb23ad376"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
