)]}'
{"id":"openstack%2Fsenlin~667215","triplet_id":"openstack%2Fsenlin~master~I3b76f1cb529a675da6ad9ba9bec3b72a66b460a1","project":"openstack/senlin","branch":"master","topic":"fix-api-ref","hashtags":[],"change_id":"I3b76f1cb529a675da6ad9ba9bec3b72a66b460a1","subject":"Fix api-ref and docs building","status":"MERGED","created":"2019-06-24 20:22:35.000000000","updated":"2019-06-28 07:09:30.000000000","submitted":"2019-06-28 07:09:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667215-1561705770593-1ef4935a","meta_rev_id":"1fcb2ab05a84d993ee2e2b64aa9ce2035f3dba58","_number":667215,"virtual_id_number":667215,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-25 05:52:55.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":2,"date":"2019-06-28 07:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-06-28 06:21:33.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-25 05:55:24.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-06-28 06:21:33.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 05:55:24.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-06-28 06:21:33.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2019-06-28 07:09:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"680993a21d2be7209f75576acf5b7ec941a8a9f9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-24 20:22:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ea89e1734d62db11d7f96dc387fbe3557bab8c5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-24 21:19:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"14563468b8552c5c5d2f763fa2baa510d9ef8990","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-24 21:20:10.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks fine, but failing, probably not related to this patch.\n\u003e dot command \u0027dot\u0027 cannot be run (needed for graphviz output), check the graphviz_dot setting","accounts_in_message":[],"_revision_number":1},{"id":"af5f712ef62ff735d962a2ee024bdb3c041fda40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 21:27:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/15/667215/1/check/build-openstack-api-ref/75e3489/html/ : SUCCESS in 2m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/667215/1/check/openstack-tox-lower-constraints/d81fa0a/ : SUCCESS in 5m 59s\n- openstack-tox-pep8 http://logs.openstack.org/15/667215/1/check/openstack-tox-pep8/eedc7cf/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/15/667215/1/check/openstack-tox-py27/4334265/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/15/667215/1/check/openstack-tox-py36/eb08f73/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/15/667215/1/check/openstack-tox-py37/e5283d3/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/15/667215/1/check/openstack-tox-docs/acf11f2/ : FAILURE in 3m 22s\n- build-openstack-releasenotes http://logs.openstack.org/15/667215/1/check/build-openstack-releasenotes/80e2983/html/ : SUCCESS in 7m 43s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/15/667215/1/check/senlin-dsvm-tempest-py27-api/4e0ab4e/ : SUCCESS in 50m 15s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/15/667215/1/check/senlin-dsvm-tempest-py3-api/a21869d/ : SUCCESS in 56m 16s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/15/667215/1/check/senlin-dsvm-tempest-py27-functional/59efc60/ : SUCCESS in 39m 41s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/15/667215/1/check/senlin-dsvm-tempest-py3-functional/4145b4d/ : SUCCESS in 37m 57s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/15/667215/1/check/senlin-dsvm-tempest-py27-integration/f6765f1/ : SUCCESS in 43m 13s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/15/667215/1/check/senlin-dsvm-tempest-py3-integration/15fe2aa/ : SUCCESS in 41m 27s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/15/667215/1/check/openstack-tox-cover/30fa0e1/cover/ : SUCCESS in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be712baecdd5f8642e5a8492977e7f4dca0e7885","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-25 05:39:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b48c1b58a8352e82e3cedcc0d783427c181475e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-25 05:52:55.000000000","message":"Patch Set 2:\n\nThe failure is indeed unrelated but we need both fixes to merge together, so I added it.","accounts_in_message":[],"_revision_number":2},{"id":"72b95a35328691d30f49657f0b6cc1f035a59e52","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-25 05:55:24.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"855f6e7d8c41a1a91000690d9ad841f46819d4e4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-25 05:55:38.000000000","message":"Patch Set 2:\n\n\u003e The failure is indeed unrelated but we need both fixes to merge\n \u003e together, so I added it.\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"b05a24bce6077151e0b4f3f28163fe5a0d051d6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 06:48:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/15/667215/2/check/build-openstack-api-ref/cc7771f/html/ : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/667215/2/check/openstack-tox-lower-constraints/083267d/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/15/667215/2/check/openstack-tox-pep8/b68ff5e/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/15/667215/2/check/openstack-tox-py27/ce5769f/ : SUCCESS in 4m 37s\n- openstack-tox-py36 http://logs.openstack.org/15/667215/2/check/openstack-tox-py36/8f752f0/ : SUCCESS in 5m 13s\n- openstack-tox-py37 http://logs.openstack.org/15/667215/2/check/openstack-tox-py37/a10c7b2/ : SUCCESS in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/15/667215/2/check/openstack-tox-docs/9a41a75/html/ : SUCCESS in 4m 06s\n- build-openstack-releasenotes http://logs.openstack.org/15/667215/2/check/build-openstack-releasenotes/fd86049/html/ : SUCCESS in 6m 58s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/15/667215/2/check/senlin-dsvm-tempest-py27-api/31688cf/ : SUCCESS in 1h 03m 00s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/15/667215/2/check/senlin-dsvm-tempest-py3-api/b4e09f9/ : SUCCESS in 45m 59s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/15/667215/2/check/senlin-dsvm-tempest-py27-functional/8acb6ba/ : SUCCESS in 45m 56s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/15/667215/2/check/senlin-dsvm-tempest-py3-functional/a5aa9d5/ : SUCCESS in 40m 52s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/15/667215/2/check/senlin-dsvm-tempest-py27-integration/12a5ba6/ : SUCCESS in 43m 35s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/15/667215/2/check/senlin-dsvm-tempest-py3-integration/3fb1c11/ : SUCCESS in 39m 11s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/15/667215/2/check/openstack-tox-cover/c063008/ : FAILURE in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e372f3d418c41fa883d8f7d619198bce90b7239d","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-06-28 06:21:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a5a2531a9cfd35fd92557904e3383a8d7827dc0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 06:21:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af60caf0aaa949ba57a8ecd7de5c94a60513abaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 07:09:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"334e94ded8a7be707646fbfa5dcd2dcf14fe3f06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 07:09:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/15/667215/2/gate/build-openstack-api-ref/ffdfd20/html/ : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/667215/2/gate/openstack-tox-lower-constraints/9135ad4/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/15/667215/2/gate/openstack-tox-pep8/b094a8d/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/15/667215/2/gate/openstack-tox-py27/43cf372/ : SUCCESS in 5m 45s\n- openstack-tox-py36 http://logs.openstack.org/15/667215/2/gate/openstack-tox-py36/ba935c1/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/15/667215/2/gate/openstack-tox-py37/a89813c/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/15/667215/2/gate/openstack-tox-docs/23daf77/html/ : SUCCESS in 4m 34s\n- build-openstack-releasenotes http://logs.openstack.org/15/667215/2/gate/build-openstack-releasenotes/ce208e1/html/ : SUCCESS in 8m 36s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/15/667215/2/gate/senlin-dsvm-tempest-py27-api/5bb645d/ : SUCCESS in 42m 18s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/15/667215/2/gate/senlin-dsvm-tempest-py27-functional/1b62e8e/ : SUCCESS in 39m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc12179412dd2ff7dcd6aef30f9e625eab020247","revisions":{"5fe049370d4f490e727fd081d54f865eace6b1df":{"kind":"REWORK","_number":1,"created":"2019-06-24 20:22:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/15/667215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/15/667215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/15/667215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/15/667215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/15/667215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/15/667215/1"}}},"commit":{"parents":[{"commit":"faf42a890b18925f522327332deb1e55ef58de78","subject":"Fix wrong assert function name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/faf42a890b18925f522327332deb1e55ef58de78"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-06-24 20:21:33.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-06-24 20:21:33.000000000","tz":120},"subject":"Fix api-ref building","message":"Fix api-ref building\n\nThe api-ref tox env is currently failing, see\nhttp://zuul.openstack.org/build/fd7de0dbca1940b08878dd62ea750fe7 for one\nexample.\n\nFix it by using the proper requirements file.\n\nChange-Id: I3b76f1cb529a675da6ad9ba9bec3b72a66b460a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/5fe049370d4f490e727fd081d54f865eace6b1df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/5fe049370d4f490e727fd081d54f865eace6b1df"}]},"branch":"refs/heads/master"},"bc12179412dd2ff7dcd6aef30f9e625eab020247":{"kind":"REWORK","_number":2,"created":"2019-06-25 05:39:29.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/15/667215/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/15/667215/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/15/667215/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/15/667215/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/15/667215/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/15/667215/2"}}},"commit":{"parents":[{"commit":"faf42a890b18925f522327332deb1e55ef58de78","subject":"Fix wrong assert function name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/faf42a890b18925f522327332deb1e55ef58de78"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-06-24 20:21:33.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-06-25 05:39:26.000000000","tz":120},"subject":"Fix api-ref and docs building","message":"Fix api-ref and docs building\n\nThe api-ref tox env is currently failing, see\nhttp://zuul.openstack.org/build/fd7de0dbca1940b08878dd62ea750fe7 for one\nexample.\n\nFix it by using the proper requirements file.\n\nAlso, add bindep.txt file so that graphviz is available. The recent\nchange to not install the global fallback broke the docs building.\n\nChange-Id: I3b76f1cb529a675da6ad9ba9bec3b72a66b460a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/bc12179412dd2ff7dcd6aef30f9e625eab020247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/bc12179412dd2ff7dcd6aef30f9e625eab020247"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
