)]}'
{"id":"openstack%2Fsenlin~669026","triplet_id":"openstack%2Fsenlin~master~I49d172414eecf9158192cf0869abb91822ee2226","project":"openstack/senlin","branch":"master","topic":"add-action-filter","hashtags":[],"change_id":"I49d172414eecf9158192cf0869abb91822ee2226","subject":"Add cluster_id to action filter in API","status":"MERGED","created":"2019-07-04 02:46:51.000000000","updated":"2019-10-30 16:58:01.000000000","submitted":"2019-10-30 16:56:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669026-1572454599422-cc9bba01","meta_rev_id":"a7ca90050c900fda0c5d9fc2060ae504e4b1916f","_number":669026,"virtual_id_number":669026,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-29 03:36:27.000000000","_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-30 16:56:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-10-30 05:44:18.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":2,"date":"2019-10-29 16:06:33.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":1,"date":"2019-10-30 16:02:41.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-11 14:05:36.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-08-14 15:10:09.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-08-14 16:10:35.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-10-30 05:44:18.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2019-10-30 16:02:41.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2019-10-30 16:56:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea1e60a48e1102a4554abb04b59f20c1d7f4230f","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-07-04 02:46:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea1fb9b808cbdd154e356719d35560636a06fd4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 03:40:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/26/669026/1/check/openstack-tox-lower-constraints/abc2fee/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/26/669026/1/check/openstack-tox-pep8/5aa23e8/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/26/669026/1/check/openstack-tox-py27/0d4c0ce/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/26/669026/1/check/openstack-tox-py36/12293ce/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/26/669026/1/check/openstack-tox-py37/16763a3/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/26/669026/1/check/openstack-tox-docs/c2a6f7c/html/ : SUCCESS in 4m 56s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/26/669026/1/check/senlin-dsvm-tempest-py27-api/606b28f/ : SUCCESS in 41m 42s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/26/669026/1/check/senlin-dsvm-tempest-py3-api/f415fb5/ : SUCCESS in 43m 21s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/26/669026/1/check/senlin-dsvm-tempest-py27-functional/d996ae6/ : SUCCESS in 41m 08s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/26/669026/1/check/senlin-dsvm-tempest-py3-functional/1ae7409/ : SUCCESS in 40m 19s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/26/669026/1/check/senlin-dsvm-tempest-py27-integration/0a2878f/ : SUCCESS in 43m 10s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/26/669026/1/check/senlin-dsvm-tempest-py3-integration/be65aee/ : SUCCESS in 47m 38s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/26/669026/1/check/openstack-tox-cover/c8a1de7/ : FAILURE in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"def579ed6508b15d98dcb13237f37d3a25a0131c","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-07-09 16:32:50.000000000","message":"Patch Set 1:\n\nThis can be a BP\nThen we do this BP:\n1API\n2tempest\n3docs\n4releaseNotes","accounts_in_message":[],"_revision_number":1},{"id":"0bd78b1a091f3760600bac812631c0386ceaa86b","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-07-10 17:18:23.000000000","message":"Patch Set 1:\n\nCan\u0027t you use the target parameter to filter out for actions belonging to a cluster?  This will give you all the cluster actions for a specific cluster.  Node actions will not be returned because the target ID is different.","accounts_in_message":[],"_revision_number":1},{"id":"e288275ae93ce19132b6424bc1baea1810abe6c1","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-07-10 17:24:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d30cc8a2c8ff2dacf5ead62e91a080958be67753","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-07-11 07:08:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"df13d56b395b572b0daaae0969f3960fbc898ff8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 08:11:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/26/669026/2/check/openstack-tox-lower-constraints/692035f/ : SUCCESS in 7m 30s\n- openstack-tox-pep8 http://logs.openstack.org/26/669026/2/check/openstack-tox-pep8/0616a5a/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/26/669026/2/check/openstack-tox-py27/c839b6a/ : SUCCESS in 6m 52s\n- openstack-tox-py36 http://logs.openstack.org/26/669026/2/check/openstack-tox-py36/6586319/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/26/669026/2/check/openstack-tox-py37/5b2dd18/ : SUCCESS in 6m 39s\n- openstack-tox-docs http://logs.openstack.org/26/669026/2/check/openstack-tox-docs/f96b0f9/html/ : SUCCESS in 4m 48s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/26/669026/2/check/senlin-dsvm-tempest-py27-api/8fad053/ : SUCCESS in 56m 40s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/26/669026/2/check/senlin-dsvm-tempest-py3-api/1c30299/ : SUCCESS in 43m 19s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/26/669026/2/check/senlin-dsvm-tempest-py27-functional/22caf21/ : SUCCESS in 36m 15s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/26/669026/2/check/senlin-dsvm-tempest-py3-functional/b20c53e/ : SUCCESS in 38m 49s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/26/669026/2/check/senlin-dsvm-tempest-py27-integration/ba85a08/ : SUCCESS in 42m 59s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/26/669026/2/check/senlin-dsvm-tempest-py3-integration/d1454b7/ : SUCCESS in 43m 49s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/26/669026/2/check/openstack-tox-cover/3d83467/ : FAILURE in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec37e9ef0d0629fafb48709b61c4d59d0d49fcbd","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-07-11 23:58:02.000000000","message":"Patch Set 2:\n\nThis looks better.  But I still have the same question about filtering using target.  Can\u0027t you use the target parameter to filter out for actions belonging to a cluster?","accounts_in_message":[],"_revision_number":2},{"id":"d4650badb1afe89402383f2aa669545e6fec9a36","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-07-12 04:10:07.000000000","message":"Patch Set 2:\n\n\u003e This looks better.  But I still have the same question about\n \u003e filtering using target.  Can\u0027t you use the target parameter to\n \u003e filter out for actions belonging to a cluster?\n\nYes, sure that!\nActually i can use target as a parameter to filter but with action have action context such as NODE_CREATE, ... i can\u0027t do. Example:\nThis a action when i do expand a cluster:\n- before:\n        {\n            \"action\": \"NODE_CREATE\",\n            \"cause\": \"Derived Action\",\n            \"created_at\": \"2019-07-12T04:02:45Z\",\n            \"data\": {},\n            \"depended_by\": [],\n            \"depends_on\": [],\n            \"end_time\": 1562904215.64605,\n            \"id\": \"fdf07488-422f-4088-80ce-0a8ed27b7b36\",\n            \"inputs\": {},\n            \"interval\": -1,\n            \"name\": \"node_create_d2efda77\",\n            \"outputs\": {},\n            \"owner\": null,\n            \"project\": \"4a92824aa74445f28d7cf814a54ee283\",\n            \"start_time\": 1562904166.23932,\n            \"status\": \"SUCCEEDED\",\n            \"status_reason\": \"Action completed successfully.\",\n            \"target\": \"d2efda77-3c83-42a1-9fc5-35a2d4f3161c\",\n            \"timeout\": 3600,\n            \"updated_at\": null,\n            \"user\": \"3502ac4dda5247829c20b6561c5d188a\"\n        },\n- after:\n        {\n            \"action\": \"NODE_CREATE\",\n            \"cause\": \"Derived Action\",\n            \"cluster_id\": \"7b1faa00-b7b8-4a21-9e32-c5f16fcc859f\",\n            \"created_at\": \"2019-07-12T04:02:45Z\",\n            \"data\": {},\n            \"depended_by\": [],\n            \"depends_on\": [],\n            \"end_time\": 1562904215.64605,\n            \"id\": \"fdf07488-422f-4088-80ce-0a8ed27b7b36\",\n            \"inputs\": {},\n            \"interval\": -1,\n            \"name\": \"node_create_d2efda77\",\n            \"outputs\": {},\n            \"owner\": null,\n            \"project\": \"4a92824aa74445f28d7cf814a54ee283\",\n            \"start_time\": 1562904166.23932,\n            \"status\": \"SUCCEEDED\",\n            \"status_reason\": \"Action completed successfully.\",\n            \"target\": \"d2efda77-3c83-42a1-9fc5-35a2d4f3161c\",\n            \"timeout\": 3600,\n            \"updated_at\": null,\n            \"user\": \"3502ac4dda5247829c20b6561c5d188a\"\n        },","accounts_in_message":[],"_revision_number":2},{"id":"7bb6dee24aa026540ba0a472c0892ea277d8c851","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-08-02 19:03:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03374502247abd0b67030c990e0c2da0efeb3ebf","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-08-02 20:14:18.000000000","message":"Patch Set 2:\n\nPlease add cluster_id to api unit test:\nhttps://github.com/openstack/senlin/blob/master/senlin/tests/unit/api/openstack/v1/test_actions.py#L88","accounts_in_message":[],"_revision_number":2},{"id":"ffa587a02513498b6ec2aebe2acb8044123ff46c","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-08-02 20:15:38.000000000","message":"Patch Set 2:\n\nAlso here: https://github.com/openstack/senlin/blob/master/senlin/tests/unit/api/openstack/v1/test_actions.py#L49","accounts_in_message":[],"_revision_number":2},{"id":"2c156610e09eec64fae6e8293588d4aaa8ba881c","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-08-02 20:43:04.000000000","message":"Patch Set 2:\n\nYou could also add cluster_id to this test:\nhttps://github.com/openstack/senlin/blob/master/senlin/tests/unit/db/shared.py#L34\n\nAnd then check for it here:\nhttps://github.com/openstack/senlin/blob/master/senlin/tests/unit/db/test_action_api.py#L46\n\nBut these tests don\u0027t really do much so this one is not super important.","accounts_in_message":[],"_revision_number":2},{"id":"8ec8d951be9ccf9905edfe4cc96f51b10b54c7f1","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-08-02 21:01:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3e929ea764603c10bec0a924ce4733f73c1b973a","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-08-02 21:19:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4c907b2a68d090e817fb9e99a062e68b0c2af0b","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-08-03 06:32:40.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\ni think when go to this step, I can absolutely return empty set and don\u0027t need query to database. And i do this was based event_list function on service.py.\n\nAnd, i think if i use get_all function, it will return all action than instead return with filter by cluster_id.","accounts_in_message":[],"_revision_number":2},{"id":"7e3434129a60cead6c8bced079650f8f74f2ac77","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-08-03 06:36:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7a0e335f4f536d43b84961821fd1196154f894cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 07:36:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/26/669026/3/check/openstack-tox-lower-constraints/a56b7bd/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://logs.opendev.org/26/669026/3/check/openstack-tox-pep8/032aaf8/ : SUCCESS in 4m 46s\n- openstack-tox-py27 https://logs.opendev.org/26/669026/3/check/openstack-tox-py27/403bc53/ : SUCCESS in 6m 13s\n- openstack-tox-py36 https://logs.opendev.org/26/669026/3/check/openstack-tox-py36/11594e7/ : SUCCESS in 5m 31s\n- openstack-tox-py37 https://logs.opendev.org/26/669026/3/check/openstack-tox-py37/7e6dcc1/ : SUCCESS in 6m 01s\n- openstack-tox-docs https://logs.opendev.org/26/669026/3/check/openstack-tox-docs/fa769bf/html/ : SUCCESS in 4m 40s\n- senlin-dsvm-tempest-py27-api https://logs.opendev.org/26/669026/3/check/senlin-dsvm-tempest-py27-api/41f4d64/ : SUCCESS in 49m 57s\n- senlin-dsvm-tempest-py3-api https://logs.opendev.org/26/669026/3/check/senlin-dsvm-tempest-py3-api/b441226/ : SUCCESS in 49m 01s\n- senlin-dsvm-tempest-py27-functional https://logs.opendev.org/26/669026/3/check/senlin-dsvm-tempest-py27-functional/30de475/ : SUCCESS in 43m 45s\n- senlin-dsvm-tempest-py3-functional https://logs.opendev.org/26/669026/3/check/senlin-dsvm-tempest-py3-functional/09713fc/ : SUCCESS in 40m 15s\n- senlin-dsvm-tempest-py27-integration https://logs.opendev.org/26/669026/3/check/senlin-dsvm-tempest-py27-integration/67676af/ : SUCCESS in 46m 45s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://logs.opendev.org/26/669026/3/check/senlin-dsvm-tempest-py3-integration/615814a/ : SUCCESS in 43m 08s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/26/669026/3/check/openstack-tox-cover/dc1d620/cover/ : SUCCESS in 6m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"34ffd17749008ea3a481cca8ff4dfaecfe42669d","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-08-11 14:05:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"19bb5a5f2119f98b0f972c225aa34a9d12b6e376","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-14 15:10:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b70278940fa98bb4a3ad63418da945e1431d56d2","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-08-14 16:10:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"df76cd5bfd9eb4052e572e03a39d73017e779450","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-09-01 14:56:10.000000000","message":"Patch Set 3: Code-Review+1\n\nBetter to update api version.","accounts_in_message":[],"_revision_number":3},{"id":"26d9744fdcd72fdec3088fb99d2d7708ee2a4357","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-21 16:43:37.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nSorry for the long wait to reply to this.  As Xuefeng said, please add microversion update as described here:\nhttps://docs.openstack.org/senlin/latest/contributor/api_microversion.html#planning-and-committing-changes\n\nAlso, please add a release note:\nhttps://docs.openstack.org/project-team-guide/release-management.html#how-to-add-new-release-notes","accounts_in_message":[],"_revision_number":3},{"id":"7e2f2908d381fa8239c57942c0aa00e475c4fe3f","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-24 08:13:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2c9395a12040da71928e33cdda43bc8320d54089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 09:15:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edca47f605d94d55b43a6adf3e2bb03c : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/418c51784fce46298170adb8506cdd5e : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0de050fccdd849a59689172d69184d9a : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/388d21a141714546ba591ca740b38a46 : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2853f8f7e9b14c568c8cf29dcdc3d3e1 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e751831522a4cd28665fa711f21409e : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70b2515c87f9460db00689133966579f : SUCCESS in 8m 41s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/29bd10f378134caa86724950fa74f652 : SUCCESS in 51m 18s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/9fcffcd78ca040bba302ffa628b5e53b : SUCCESS in 54m 10s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/0b2bf82d7048495491884f4ea11646fc : SUCCESS in 50m 52s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/5d3bf6e7518b4dfab4c7f45a6886a021 : SUCCESS in 39m 39s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/58fdfd4f84d249c8aefb766d8b34839d : SUCCESS in 49m 34s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/feb08905b0024445b99cef1ab76defbe : FAILURE in 43m 24s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/245d1c56029b48f387b24995e72035eb : FAILURE in 42m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe5225e557344dd4b2e90c48f869df50 : FAILURE in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"01a3e658fbbef1a2fba6427e860c4ab775d3ac21","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-24 17:00:18.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f9bba5193400627699c086b6d52acde51b25ab7d","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-25 02:22:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a5217763913950c4e67f5cd6273760d71461bf55","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-25 02:22:41.000000000","message":"Patch Set 4:\n\n\u003e (6 comments)\n\nthank you for your advice","accounts_in_message":[],"_revision_number":4},{"id":"11b3d350ca416a6b2699c62cc3e3a937354b114b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 03:21:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb816ebc41e84e6cad5b42651707d7bb : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dc8889fc1e4468b98472c6c80d0e93d : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f23e46dd5b9045fe8ef7b8104e88e155 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7d3eed22f7a43b9b1661b0dc8fd2d73 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da5b16dbd6c34488bfd30f2c9dcccbe8 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/059fea7707ab4052be5e6dfa5232d7c2 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebb6a0ce337549cab38c96f9fdcd5749 : SUCCESS in 10m 09s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/309ff3f6ec6344db99e30c2bfe9275ff : SUCCESS in 45m 48s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/76669003523c4378a2c95912f9113475 : SUCCESS in 53m 20s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/d52306495d964d69aec863a20a886047 : SUCCESS in 43m 49s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/eaf9b8581e434b1fb49f45ddd623d6ab : SUCCESS in 44m 38s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/894534e41b784167a5b5a963086d8e47 : SUCCESS in 43m 51s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/dc5c9b0dc79d434897b96c0c96f60615 : FAILURE in 47m 56s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/ffcbbaee50fe4c60bcbb17ca8d94e708 : FAILURE in 50m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93f4f9df8f314179a73d052d9eb572a4 : FAILURE in 8m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd0d9f5c3100d76df472aa09cbdf329a3ed220a1","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-25 16:39:53.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nCan you also change senlin/tests/unit/api/openstack/v1/test_action.py to test that cluster_id is only returned for microversion \u003e 1.13\nE.g. https://github.com/openstack/senlin/blob/master/senlin/tests/unit/api/openstack/v1/test_nodes.py#L82","accounts_in_message":[],"_revision_number":5},{"id":"eea34ad16c30e912ca1c5f45e7542df70deba0c2","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-26 02:26:10.000000000","message":"Patch Set 5:\n\n\u003e (2 comments)\n \u003e \n \u003e Can you also change senlin/tests/unit/api/openstack/v1/test_action.py\n \u003e to test that cluster_id is only returned for microversion \u003e 1.13\n \u003e E.g. https://github.com/openstack/senlin/blob/master/senlin/tests/unit/api/openstack/v1/test_nodes.py#L82\n\nyep, thank you very much","accounts_in_message":[],"_revision_number":5},{"id":"e0a15c28bc913ae54712b94cd51a7378968ed382","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-26 02:27:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c3a99eadb85acd0eefba4b254ac977bb7fcee491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 03:27:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/908cd5c292814c31ad4cc77153485ca5 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e69c179337fa442f978dd3d08247b368 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c19d34229ca7431eb3754bddbb087557 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a8bba1ae59a4a3a847380c9ab1cb1f9 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55ef3e0f46fa48a88ff6039b74461561 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c1bdaf040d41ddb7561aa4c8f470fb : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b18a4060c7244618eb9f83e82762d3c : SUCCESS in 8m 48s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/34f1cf4218cd401e9e7d16df1c8ee8db : SUCCESS in 47m 01s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/5773e36511c146388951fcfb9911e87d : SUCCESS in 48m 33s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/4f0463f1fd434170887cc53a99196ad4 : SUCCESS in 42m 53s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/75a2719437ea44098073d1637b1e7761 : SUCCESS in 39m 49s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/e4e64c716b234ecd81fca498cbdfcfa6 : SUCCESS in 37m 16s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/d981648b8d114adcb49df0e4bc0e9290 : FAILURE in 45m 54s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/ac4ff0a0a3ce489992b6254ed7a27ee6 : FAILURE in 40m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58d79a747e1944599b9056c6c76915cf : FAILURE in 8m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"699a07dc1e75e1ee971faba52da7bdafaec1178b","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-28 16:32:33.000000000","message":"Patch Set 6:\n\nCan you also add a test to check that cluster_id is returned for 1.14 or greater?  Something like this:\nhttps://github.com/openstack/senlin/blob/master/senlin/tests/unit/api/openstack/v1/test_nodes.py#L126\n\nThat\u0027s the only thing I see missing.","accounts_in_message":[],"_revision_number":6},{"id":"1d3eb6bf6527733824244f6de5d602f703aabd22","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-29 03:35:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ef24d63108f43e3d7d00cadb0e8b7abe4d816162","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-10-29 03:36:27.000000000","message":"Patch Set 7:\n\n\u003e Can you also add a test to check that cluster_id is returned for\n \u003e 1.14 or greater?  Something like this:\n \u003e https://github.com/openstack/senlin/blob/master/senlin/tests/unit/api/openstack/v1/test_nodes.py#L126\n \u003e \n \u003e That\u0027s the only thing I see missing.\n\nYep, thank you.","accounts_in_message":[],"_revision_number":7},{"id":"e3c73b1b916f2747d5fe5d1efe20f80e68751651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 04:27:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cab932c3288f498baedd60746588ebf8 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b031def0d2ae4cecb603573c21a9f096 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2db0d4c72d244b58b9dee9ace0c3983c : SUCCESS in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2381a66ac9345c292317a5653472a8a : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13668912fa234ce98b44d885f2b50b94 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb2b7385f2324e379fff65eaee006e19 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5d42694603945fcbf3eec7bcf073c43 : SUCCESS in 12m 22s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/1f2c373e5ad340ff85b3650e3ca27464 : SUCCESS in 40m 22s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/1a70ff058157458694bb633c50ffe67e : SUCCESS in 41m 43s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8f3f95f1c31243ffb72ce78694b7496d : SUCCESS in 43m 39s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/5ac416c58e93460f89a36fc5469c6512 : SUCCESS in 36m 08s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/60775a23fe3e4f9bb4977bd7fa058c43 : SUCCESS in 38m 02s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/6f7b12f8d26c4f3c83e7a102c8ef4d8f : FAILURE in 43m 06s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/d05dd26d5a24411e9bb32c492952236b : FAILURE in 45m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27658dc4993d4deda9df754185b95490 : FAILURE in 7m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7429af128017b62701d137bb9bf85af579c08a49","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-29 16:06:33.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM.  Thanks for making all the requested changes.","accounts_in_message":[],"_revision_number":7},{"id":"1dee8f351704737e8b4e8c02a2622d3117d22dec","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-10-30 05:44:18.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":7},{"id":"c19b063b63b15751279554ac42a12eeac929dd93","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-30 16:02:41.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e6404e4e14c06b0fca3aeae4bb3da3796a7dffeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 16:02:56.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"bad6b59c9a069ae43e157d91a09204d09ce0db68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 16:56:39.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73898c480e754235ac7e3f5b074235fc : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddede61619fb4c2a94407ffb43509887 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5b2514c5895439b9ad801f95bad184d : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26e0b01e5eaa43dfb4b572f1944eb741 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41cf24cb0dc247e6a4dbe327aeab33ec : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b5898d91e84e119aa98f72c72b87e8 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c9d94521ca640ff845a9cd966dd40f7 : SUCCESS in 8m 52s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/3f3dcb05de4d4ba68ce1b0d9bccda138 : SUCCESS in 40m 05s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/bc493223246842efb1998856c1a399d4 : SUCCESS in 46m 32s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/56ef4aed7c7749eab1bd5b79aafac6c0 : SUCCESS in 52m 05s","accounts_in_message":[],"_revision_number":7},{"id":"3c08106749afbee96d535568d3063c42539c5f45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 16:56:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"5e7f4b379c78800b63ba31600b5bc5f9e1892ff4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 16:58:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea6a86a7563149f88a1a1d3813fbf0e0 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0057ed4d702648c089d4af1ef625efec : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8a5460c591783d848622c9e54c88aedb4458c935","revisions":{"e2da59f58f2971f823ec4799af13e2007988a01c":{"kind":"REWORK","_number":1,"created":"2019-07-04 02:46:51.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/26/669026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/26/669026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/26/669026/1"}}},"commit":{"parents":[{"commit":"bc12179412dd2ff7dcd6aef30f9e625eab020247","subject":"Fix api-ref and docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/bc12179412dd2ff7dcd6aef30f9e625eab020247"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-07-04 02:33:43.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-07-04 02:33:43.000000000","tz":420},"subject":"Add cluster_id to action filter in API","message":"Add cluster_id to action filter in API\n\nCurrent, an action was created haven\u0027t information of a cluster.\nWhen a specific action is created, we do not know what action this action takes\nin a cluster and when we execute the query API:\n\nGET http://senlin_host:8777/actions\n\nso many actions are returned even though we can limit using parameters\nsuch as name, target, action, status combined with limit and marker.\nBut, we can\u0027t just get the actions executed in a particular cluster.\nSo, I want us to be able to do this by enabling the query with a cluster_id parameter.\nTo accomplish this, I need to modify the schema of\naction models and action tables in the database of senlin service.\n\nChange-Id: I49d172414eecf9158192cf0869abb91822ee2226\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e2da59f58f2971f823ec4799af13e2007988a01c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e2da59f58f2971f823ec4799af13e2007988a01c"}]},"branch":"refs/heads/master"},"e5274ce14a4c036c52282bd7b85512191638dcfd":{"kind":"REWORK","_number":2,"created":"2019-07-11 07:08:57.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/26/669026/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/26/669026/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/26/669026/2"}}},"commit":{"parents":[{"commit":"bc12179412dd2ff7dcd6aef30f9e625eab020247","subject":"Fix api-ref and docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/bc12179412dd2ff7dcd6aef30f9e625eab020247"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-07-04 02:33:43.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-07-11 07:08:22.000000000","tz":420},"subject":"Add cluster_id to action filter in API","message":"Add cluster_id to action filter in API\n\nCurrent, an action was created haven\u0027t information of a cluster.\nWhen a specific action is created, we do not know what action this action takes\nin a cluster and when we execute the query API:\n\nGET http://senlin_host:8777/actions\n\nso many actions are returned even though we can limit using parameters\nsuch as name, target, action, status combined with limit and marker.\nBut, we can\u0027t just get the actions executed in a particular cluster.\nSo, I want us to be able to do this by enabling the query with a cluster_id parameter.\nTo accomplish this, I need to modify the schema of\naction models and action tables in the database of senlin service.\n\nChange-Id: I49d172414eecf9158192cf0869abb91822ee2226\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e5274ce14a4c036c52282bd7b85512191638dcfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e5274ce14a4c036c52282bd7b85512191638dcfd"}]},"branch":"refs/heads/master"},"12ced06695bbc997cec756ddd58593c8d25a022b":{"kind":"REWORK","_number":3,"created":"2019-08-03 06:36:10.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/26/669026/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/26/669026/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/26/669026/3"}}},"commit":{"parents":[{"commit":"bc12179412dd2ff7dcd6aef30f9e625eab020247","subject":"Fix api-ref and docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/bc12179412dd2ff7dcd6aef30f9e625eab020247"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-07-04 02:33:43.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-08-03 06:35:43.000000000","tz":420},"subject":"Add cluster_id to action filter in API","message":"Add cluster_id to action filter in API\n\nCurrent, an action was created haven\u0027t information of a cluster.\nWhen a specific action is created, we do not know what action this action takes\nin a cluster and when we execute the query API:\n\nGET http://senlin_host:8777/actions\n\nso many actions are returned even though we can limit using parameters\nsuch as name, target, action, status combined with limit and marker.\nBut, we can\u0027t just get the actions executed in a particular cluster.\nSo, I want us to be able to do this by enabling the query with a cluster_id parameter.\nTo accomplish this, I need to modify the schema of\naction models and action tables in the database of senlin service.\n\nChange-Id: I49d172414eecf9158192cf0869abb91822ee2226\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/12ced06695bbc997cec756ddd58593c8d25a022b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/12ced06695bbc997cec756ddd58593c8d25a022b"}]},"branch":"refs/heads/master"},"16ce848ad15b6289ba43e3f7ecc7405073255058":{"kind":"REWORK","_number":4,"created":"2019-10-24 08:13:50.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/26/669026/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/26/669026/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/26/669026/4"}}},"commit":{"parents":[{"commit":"ac493538991b5ea0a06bf7904efe9ebc4bf7333b","subject":"Merge \"Add tainted field to nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ac493538991b5ea0a06bf7904efe9ebc4bf7333b"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-07-04 02:33:43.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-10-24 08:13:18.000000000","tz":420},"subject":"Add cluster_id to action filter in API","message":"Add cluster_id to action filter in API\n\nCurrent, an action was created haven\u0027t information of a cluster.\nWhen a specific action is created, we do not know what action this action takes\nin a cluster and when we execute the query API:\n\nGET http://senlin_host:8777/actions\n\nso many actions are returned even though we can limit using parameters\nsuch as name, target, action, status combined with limit and marker.\nBut, we can\u0027t just get the actions executed in a particular cluster.\nSo, I want us to be able to do this by enabling the query with a cluster_id parameter.\nTo accomplish this, I need to modify the schema of\naction models and action tables in the database of senlin service.\n\nChange-Id: I49d172414eecf9158192cf0869abb91822ee2226\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/16ce848ad15b6289ba43e3f7ecc7405073255058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/16ce848ad15b6289ba43e3f7ecc7405073255058"}]},"branch":"refs/heads/master"},"c25dfd597cd624a8dd537b051fb8398ef90671e1":{"kind":"REWORK","_number":5,"created":"2019-10-25 02:22:08.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/26/669026/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/26/669026/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/26/669026/5"}}},"commit":{"parents":[{"commit":"ac493538991b5ea0a06bf7904efe9ebc4bf7333b","subject":"Merge \"Add tainted field to nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ac493538991b5ea0a06bf7904efe9ebc4bf7333b"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-07-04 02:33:43.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-10-25 02:21:55.000000000","tz":420},"subject":"Add cluster_id to action filter in API","message":"Add cluster_id to action filter in API\n\nCurrent, an action was created haven\u0027t information of a cluster.\nWhen a specific action is created, we do not know what action this action takes\nin a cluster and when we execute the query API:\n\nGET http://senlin_host:8777/actions\n\nso many actions are returned even though we can limit using parameters\nsuch as name, target, action, status combined with limit and marker.\nBut, we can\u0027t just get the actions executed in a particular cluster.\nSo, I want us to be able to do this by enabling the query with a cluster_id parameter.\nTo accomplish this, I need to modify the schema of\naction models and action tables in the database of senlin service.\n\nChange-Id: I49d172414eecf9158192cf0869abb91822ee2226\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c25dfd597cd624a8dd537b051fb8398ef90671e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c25dfd597cd624a8dd537b051fb8398ef90671e1"}]},"branch":"refs/heads/master"},"d515c183e92a6b076a5562e456411edd014e3355":{"kind":"REWORK","_number":6,"created":"2019-10-26 02:27:24.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/26/669026/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/26/669026/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/26/669026/6"}}},"commit":{"parents":[{"commit":"ac493538991b5ea0a06bf7904efe9ebc4bf7333b","subject":"Merge \"Add tainted field to nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ac493538991b5ea0a06bf7904efe9ebc4bf7333b"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-07-04 02:33:43.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-10-26 02:26:32.000000000","tz":420},"subject":"Add cluster_id to action filter in API","message":"Add cluster_id to action filter in API\n\nCurrent, an action was created haven\u0027t information of a cluster.\nWhen a specific action is created, we do not know what action this action takes\nin a cluster and when we execute the query API:\n\nGET http://senlin_host:8777/actions\n\nso many actions are returned even though we can limit using parameters\nsuch as name, target, action, status combined with limit and marker.\nBut, we can\u0027t just get the actions executed in a particular cluster.\nSo, I want us to be able to do this by enabling the query with a cluster_id parameter.\nTo accomplish this, I need to modify the schema of\naction models and action tables in the database of senlin service.\n\nChange-Id: I49d172414eecf9158192cf0869abb91822ee2226\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d515c183e92a6b076a5562e456411edd014e3355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d515c183e92a6b076a5562e456411edd014e3355"}]},"branch":"refs/heads/master"},"8a5460c591783d848622c9e54c88aedb4458c935":{"kind":"REWORK","_number":7,"created":"2019-10-29 03:35:38.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/26/669026/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/26/669026/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/26/669026/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/26/669026/7"}}},"commit":{"parents":[{"commit":"ac493538991b5ea0a06bf7904efe9ebc4bf7333b","subject":"Merge \"Add tainted field to nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ac493538991b5ea0a06bf7904efe9ebc4bf7333b"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-07-04 02:33:43.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-10-29 03:35:22.000000000","tz":420},"subject":"Add cluster_id to action filter in API","message":"Add cluster_id to action filter in API\n\nCurrent, an action was created haven\u0027t information of a cluster.\nWhen a specific action is created, we do not know what action this action takes\nin a cluster and when we execute the query API:\n\nGET http://senlin_host:8777/actions\n\nso many actions are returned even though we can limit using parameters\nsuch as name, target, action, status combined with limit and marker.\nBut, we can\u0027t just get the actions executed in a particular cluster.\nSo, I want us to be able to do this by enabling the query with a cluster_id parameter.\nTo accomplish this, I need to modify the schema of\naction models and action tables in the database of senlin service.\n\nChange-Id: I49d172414eecf9158192cf0869abb91822ee2226\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/8a5460c591783d848622c9e54c88aedb4458c935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/8a5460c591783d848622c9e54c88aedb4458c935"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
