)]}'
{"id":"openstack%2Fsenlin~669505","triplet_id":"openstack%2Fsenlin~master~Ie33b1b1916f99d42d4d091a9532b8608f1a6fc47","project":"openstack/senlin","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ie33b1b1916f99d42d4d091a9532b8608f1a6fc47","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 20:12:19.000000000","updated":"2019-07-10 17:47:58.000000000","submitted":"2019-07-10 17:47:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669505-1562780878600-b7e4bc0b","meta_rev_id":"f5e0ae04a800f47d4ffef8634d34d6efd7149c5b","_number":669505,"virtual_id_number":669505,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":2,"date":"2019-07-10 17:47:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":2,"date":"2019-07-09 16:35:39.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":2,"date":"2019-07-10 16:55:28.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"all":[{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":1,"date":"2019-07-10 16:55:28.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 16:35:39.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2019-07-10 16:55:28.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2019-07-10 17:47:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b21e32f29b91465c15ac658090a47d1075dd770e","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 20:12:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6dd5eb308d43809bd9f72eea0960df1c068a2a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:04:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/05/669505/1/check/build-openstack-api-ref/3344bb3/html/ : SUCCESS in 3m 12s\n- test-release-openstack http://logs.openstack.org/05/669505/1/check/test-release-openstack/6b721b9/ : SUCCESS in 2m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/669505/1/check/openstack-tox-lower-constraints/4df1910/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/05/669505/1/check/openstack-tox-pep8/58056ae/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/05/669505/1/check/openstack-tox-py27/c52a094/ : SUCCESS in 6m 52s\n- openstack-tox-py36 http://logs.openstack.org/05/669505/1/check/openstack-tox-py36/5759f40/ : SUCCESS in 5m 08s\n- openstack-tox-py37 http://logs.openstack.org/05/669505/1/check/openstack-tox-py37/e9cd232/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/05/669505/1/check/openstack-tox-docs/afcaa81/html/ : SUCCESS in 4m 18s\n- build-openstack-releasenotes http://logs.openstack.org/05/669505/1/check/build-openstack-releasenotes/4e043b6/html/ : SUCCESS in 7m 31s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/05/669505/1/check/senlin-dsvm-tempest-py27-api/0750355/ : SUCCESS in 44m 17s\n- senlin-dsvm-tempest-py3-api http://logs.openstack.org/05/669505/1/check/senlin-dsvm-tempest-py3-api/accb1a0/ : SUCCESS in 49m 14s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/05/669505/1/check/senlin-dsvm-tempest-py27-functional/b504fc4/ : SUCCESS in 37m 19s\n- senlin-dsvm-tempest-py3-functional http://logs.openstack.org/05/669505/1/check/senlin-dsvm-tempest-py3-functional/65d0204/ : SUCCESS in 36m 58s\n- senlin-dsvm-tempest-py27-integration http://logs.openstack.org/05/669505/1/check/senlin-dsvm-tempest-py27-integration/a7bccd2/ : SUCCESS in 43m 24s (non-voting)\n- senlin-dsvm-tempest-py3-integration http://logs.openstack.org/05/669505/1/check/senlin-dsvm-tempest-py3-integration/4d4d5de/ : SUCCESS in 40m 24s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/05/669505/1/check/openstack-tox-cover/0566236/cover/ : SUCCESS in 6m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97c43bf2521e4f0ec617051e4110872d508cd295","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-07-09 16:35:39.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"7bb584deb91d59fdf753c4b0a92be56834d98ebb","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-07-10 16:55:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d38dee1ce42fa8714ed15eeec1483f5abe4173b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:55:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfee94c1f55f7afdeebc7bb124359179107661e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:47:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/05/669505/1/gate/build-openstack-api-ref/4328ca9/html/ : SUCCESS in 3m 00s\n- test-release-openstack http://logs.openstack.org/05/669505/1/gate/test-release-openstack/bfad8f8/ : SUCCESS in 3m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/669505/1/gate/openstack-tox-lower-constraints/581a1c0/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/05/669505/1/gate/openstack-tox-pep8/bc57215/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/05/669505/1/gate/openstack-tox-py27/07cd1a5/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/05/669505/1/gate/openstack-tox-py36/4353e32/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/05/669505/1/gate/openstack-tox-py37/27a5660/ : SUCCESS in 5m 07s\n- openstack-tox-docs http://logs.openstack.org/05/669505/1/gate/openstack-tox-docs/58afb8a/html/ : SUCCESS in 3m 40s\n- build-openstack-releasenotes http://logs.openstack.org/05/669505/1/gate/build-openstack-releasenotes/554cf7a/html/ : SUCCESS in 8m 13s\n- senlin-dsvm-tempest-py27-api http://logs.openstack.org/05/669505/1/gate/senlin-dsvm-tempest-py27-api/c0474b5/ : SUCCESS in 42m 40s\n- senlin-dsvm-tempest-py27-functional http://logs.openstack.org/05/669505/1/gate/senlin-dsvm-tempest-py27-functional/e7c3d39/ : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":1},{"id":"30e7cbd5151f1525fb44ca4326dff7ef90c431f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:47:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08a32789d32e173f1d7fade353f69d0082b54aa3","revisions":{"08a32789d32e173f1d7fade353f69d0082b54aa3":{"kind":"REWORK","_number":1,"created":"2019-07-05 20:12:19.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/05/669505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/05/669505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/05/669505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/05/669505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/05/669505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/05/669505/1"}}},"commit":{"parents":[{"commit":"bc12179412dd2ff7dcd6aef30f9e625eab020247","subject":"Fix api-ref and docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/bc12179412dd2ff7dcd6aef30f9e625eab020247"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:01:20.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 20:12:18.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ie33b1b1916f99d42d4d091a9532b8608f1a6fc47\nStory: #2005924\nTask: #34246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/08a32789d32e173f1d7fade353f69d0082b54aa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/08a32789d32e173f1d7fade353f69d0082b54aa3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
