)]}'
{"id":"openstack%2Fsenlin~676910","triplet_id":"openstack%2Fsenlin~master~Ib7695212e42ef16dbc1edd65c3516b277e99078d","project":"openstack/senlin","branch":"master","topic":"ipv6-only-deployment-and-testing","hashtags":[],"change_id":"Ib7695212e42ef16dbc1edd65c3516b277e99078d","subject":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate","status":"MERGED","created":"2019-08-16 11:32:42.000000000","updated":"2019-10-23 01:51:48.000000000","submitted":"2019-10-23 01:49:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676910-1571795362542-ed58b23b","meta_rev_id":"d79360c93d4eaca3725f81de9fee245ee1d08259","_number":676910,"virtual_id_number":676910,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 01:49:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-22 13:08:44.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"all":[{"value":2,"date":"2019-10-22 22:13:33.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-10-21 20:45:56.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"all":[{"value":1,"date":"2019-10-22 22:13:33.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"date":"2019-10-21 20:46:20.000000000","_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 20:45:56.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-10-21 20:46:20.000000000","updated_by":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"reviewer":{"_account_id":25674,"name":"Jude Cross","email":"jucross@blizzard.com","username":"problem_v"},"state":"REVIEWER"},{"updated":"2019-10-22 22:13:33.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2019-10-23 01:49:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"209d8da8b0acc4608a07a4f17d56b64bf2f0da8f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-16 11:32:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4947dcce96d61be227e00f7593d522cd3c8bb668","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:27:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f57fff2353d4f49afc76f33281019cc : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/766a51a88d18448f94c12ab7e90e1f57 : SUCCESS in 8m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5038e5c54fef40c2b5685b0e8841e4a2 : SUCCESS in 10m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a406a6bced04793849613a9b80bdb52 : SUCCESS in 10m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64236f59caa4408e9d4f2a837eafe131 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6af9e0b566402a97ed959dff96c734 : SUCCESS in 5m 54s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/35cdc8e1f89f44cfbd63aa2c9b09632d : SUCCESS in 44m 52s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/864c159c2d3a43f19197c6b468bde84b : SUCCESS in 48m 54s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/5fcdca5394754ef680e3ea43f5add892 : FAILURE in 5m 25s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/edfbe52518dc40a2b584c24a1ab6bb9e : SUCCESS in 41m 13s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/2abb1e861c5043eaac03e44d16c3967e : SUCCESS in 41m 30s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/b67ac0f2623f4fc8a6d4028f04342dfe : SUCCESS in 46m 54s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/e86b374f674c4da3b7772dbe6d8694b8 : SUCCESS in 40m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5575ae68a9794d56a2d45f33bee3e34b : SUCCESS in 7m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7c0d1c81be2b80af68e207a4ced247a541b2802","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-16 12:40:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c6adf309def1de60f5f4b7c2138d6974c950b60c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 13:33:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c767f7e7b3d4af1a9f0f7f48845761f : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/169e8d1203d7467d8daa4a233493d6c9 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a223e1d786fe445aa48e4362087c0edd : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51c01bfc683a495595aa987262b69888 : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44eadeece3694c9fb440ff0dcb98c535 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74db24a76ee49f994aa70800da592dc : SUCCESS in 5m 58s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/72a69892a14d489ba087c9c2f5c5bb42 : SUCCESS in 46m 37s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/92d3dd15dd3d49da8f0f785ea8f7b7db : SUCCESS in 44m 49s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/a9546666598e4531b10004ad804012c6 : FAILURE in 7m 13s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/f66a5a51a8584f1080a86eee6484c44b : SUCCESS in 41m 17s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/89712f28dc9047e180b4b98efe6f75e0 : SUCCESS in 40m 01s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/9d6e7e7e6cc741e49e943bc330c14f6e : SUCCESS in 42m 26s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/0bb326bc78a34d558c9f5dad307d2830 : SUCCESS in 49m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a57999cdf3d471684c12fe0c6d56dfd : SUCCESS in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97b97cb49fd310ab210d639b99c0c98a30a69bf1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-16 18:07:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d7fd6c5576fc88b90475686844366031654c50aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 19:07:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d2b51da4bed4449bfcfdc7e655e5cea : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f01a29a5a44770ad96f37561b67534 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4aa8aa8ba1d74a08aee8a25fe9a596aa : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f09d721df36495eb13213d7e7d1d652 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f9b94c262b245abb53a355850d836bf : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9bb04d41474d82876c8ecb3aa9d28b : SUCCESS in 4m 18s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/6098d026b3ea4e53a6b609daccc0211b : SUCCESS in 41m 04s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/f80e8bd29a8f4c0384c043d0cd91c95a : SUCCESS in 51m 55s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/ebcb8930614c4f27b67f480d0091b55d : FAILURE in 30m 25s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/df3b2827fbc14e65b91cc26dfb5b7a02 : SUCCESS in 58m 52s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/f5ee705edcf04281915ad37bca31a9ad : SUCCESS in 40m 33s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/2e6a53590da34701868091f5794e44bc : SUCCESS in 41m 08s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/31c92f233ca1423b834016008b393853 : SUCCESS in 43m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01a0e3db75944059bb827627bb64c0c2 : SUCCESS in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d50aa4f855553b33a4595edf02f4f5c3f5d2724","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-18 14:15:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6a01364f644ab45d08a750790a2897771e349699","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 15:05:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbba7131934f40cfb2d60cd6ab767c83 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/775f1c5384a04d69b3bdd16144ca1f48 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8656e24edac24ea38798cd991c0d8bf6 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/791a82aff73b4145b62836feba6ad493 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bce9c4e95e5247eb89e5d6bdf9543e37 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e1cde6e209d4e6b9eb47dc91cec189b : SUCCESS in 3m 58s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/2bc0ee1a626742719861e575c83ecb8e : SUCCESS in 43m 49s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/36fc8fabf0714c5a82f87b8a77ff8803 : SUCCESS in 44m 25s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/0ad3b4aac0424ad78171ca7546421f5e : FAILURE in 37m 53s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/04e7e785e64f46c0a98804a29589c827 : SUCCESS in 37m 32s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/47ad935ee6b44a109658afe2e56ed060 : SUCCESS in 44m 12s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/4fc7ad9f915f4eceb9d5393a86addfff : SUCCESS in 42m 24s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/32ee986960d54707bf600b8f31ff9fd4 : SUCCESS in 39m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e3efd9366494d5b9c25be7c5e060ab9 : SUCCESS in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f6c638ff53d333709808c071054add8582c60727","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-18 05:55:36.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"600731d5816ffb997bbb58bb7560a8415eab1a63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 05:55:50.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"bbfc3de22b367fe8781a81aed56e71407dcd44d1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-18 05:56:09.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"10a4f7de8ee8336005f52343d32e6d26960af852","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-18 05:59:19.000000000","message":"Patch Set 5:\n\nCan you change this to just be non-voting?","accounts_in_message":[],"_revision_number":5},{"id":"4856d0dc8ddd7f14161e27c24c1bcb157cfcbbba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 07:32:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4ee131df1bc43afad6893f530b1d1bb : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d5d958fd0c149a98e76013bcdf358b0 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3bdf3ae4d154128809a6f7276614d6d : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7262c6d02914b05bf785904e2d8c4d3 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a8cbe7f52cb452185e2039c7718b524 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4069fb260ced4bc28984a24a10e6d908 : SUCCESS in 3m 59s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/92985335c1824b36b350658b25de539a : SUCCESS in 41m 51s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/efe8d8f39e4b46fc8b5bb84d2c304818 : SUCCESS in 53m 13s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/c9c03d3a1ae24a9a874d8016a00a191c : FAILURE in 39m 23s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/87f9d565078243658e38d83712a0417d : SUCCESS in 41m 23s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/0901a7f2c37c419ba0e38c7a7b5dc972 : SUCCESS in 45m 38s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/5b9f6f53924141d68bf52694acad6d96 : SUCCESS in 48m 18s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/87d37084688543d3966fcbcc6870552b : SUCCESS in 52m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3635f78cd0e64131a4604307f26249a5 : SUCCESS in 6m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"026475824ad84466331736ea7683b462113948ff","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-21 01:29:22.000000000","message":"Patch Set 5:\n\n\u003e Can you change this to just be non-voting?\n\nactually as per goal, we need to make it voting job but how long you think the fix can take ?","accounts_in_message":[],"_revision_number":5},{"id":"e511579058a2263e87239a75936573f05f733529","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-21 03:06:36.000000000","message":"Patch Set 5:\n\n\u003e \u003e Can you change this to just be non-voting?\n \u003e \n \u003e actually as per goal, we need to make it voting job but how long\n \u003e you think the fix can take ?\n\nhttps://review.opendev.org/#/c/689315/","accounts_in_message":[],"_revision_number":5},{"id":"ac27636cdf134e21bcdad78a5b3b83fe483fe1c5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-21 18:43:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fed493cc3ddf8e11bc57c3e7e2c15ff786b24c2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 19:37:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d41c31dfc8c4d2d96c96b8545424505 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aacaee2915c4c4ba6012fe7a1d2a8f9 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8eab591ab6d24b9bad97333940332422 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89dece560bea4207bd3537a25daa0ed0 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a44edc2629c049798ccdf8a96eb3dbc1 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0247fd6a7a28479ea4ff9f2a98d36a43 : SUCCESS in 4m 08s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/042cbae46a8b4d0bb7ad0739456293ba : SUCCESS in 47m 19s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/509f3b98fe6747f38c5a50eaa16ff68a : SUCCESS in 44m 14s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/deb31d6c459944f6b3a8b275f6f5b1de : FAILURE in 34m 55s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/2a914a5cfd504e75a7fb9c744eade2d9 : SUCCESS in 38m 01s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/51929bb97e104de29ceb44dbaa6d55bb : SUCCESS in 45m 15s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/4cbb94e65c3240ffbe8c8b69bd5dad4c : FAILURE in 45m 44s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/ab90c3f47c6d409b9e230fae11f4a0e9 : SUCCESS in 47m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99256a43bbb7489a8bde0e3b3a598ebc : SUCCESS in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b359a350755c9a6cec8eae1d1a6b5e3bbfafb272","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-21 19:59:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"93e52cdb7c7681a74c4f410bc9ff119eedb7a516","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-21 20:45:56.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"0278ae807588337b50774b0fcdbec821ae5273e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 20:58:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f165213db5a4165a284730f2e4716ab : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c79c4cd0937419faf6632fd3317f019 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fcad31bc9651450285174bf3adb48236 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f73350263d9e49c289ae18088f38056c : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79e36444910149fa82d2efe4d277c2dc : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a7051e128e34b37a14a10a52851b4ab : SUCCESS in 4m 22s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/ac7d55e66f2143e88e96fc0d1122f54e : SUCCESS in 48m 34s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/d33da6618e044ab8be5348aa349667b5 : SUCCESS in 53m 18s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/945985f5729549ad8abdff7210ff61ed : SUCCESS in 48m 28s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/e3375f29b2014574a0b48aa7375c6bb8 : SUCCESS in 41m 11s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/2bd47189ed0e4181b773edac59edbb41 : SUCCESS in 37m 43s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/212cf8cf1c6d4d7481d47d716454b31f : FAILURE in 48m 24s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/95e48ff6e8924bc398e30fabbf7c0f1b : FAILURE in 55m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c13c4e1cace45eb8f38fd94eb3d6c2c : SUCCESS in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c82f8f51b3bc940176d853d1342a391c3757a195","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-21 21:08:36.000000000","message":"Patch Set 1:\n\nIs there a reason why we only run the API tests and not the functional and integration tests with ipv6?","accounts_in_message":[],"_revision_number":1},{"id":"066f16be96958f8f578eff1c2740f09a49ad18b9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-22 13:08:44.000000000","message":"Patch Set 7:\n\n\u003e Is there a reason why we only run the API tests and not the\n \u003e functional and integration tests with ipv6?\n\nWe can do that if you think that is useful to test from IP version perspective. We can always extend this job test scope as per need. For example if we add any IPv6 specific integration test then we can add the same to run as part of this job. \n\nI can extend this to add integration and functional tests also if needed ?","accounts_in_message":[],"_revision_number":7},{"id":"a2d641a2228113cdff2f912bf016d3c612c2795d","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-22 22:13:33.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n\u003e We can do that if you think that is useful to test from IP version\n \u003e perspective. We can always extend this job test scope as per need.\n \u003e For example if we add any IPv6 specific integration test then we\n \u003e can add the same to run as part of this job.\n \u003e \n \u003e I can extend this to add integration and functional tests also if\n \u003e needed ?\n\nOk, we can leave it as is for now since we don\u0027t really have IPv6 specific tests.  LGTM.","accounts_in_message":[],"_revision_number":7},{"id":"ac991bcfd66bf5601b00b6467d608f9157a58998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 22:13:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9e15cb0ad72e7f0150d28d2962ae5de4c2a76a8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 23:11:16.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd9dc36f26ef497398eb7e58fd94a5f4 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618b68bf95584a84bb004fac863d2eeb : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23cc9e07360e498ca2357fb126c3db71 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f207a97115bf4e1380d0dd5152ec901b : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bf58ed45cb040a6b955ccd264373cf0 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522d0ddc76a740b299ddc793fcb7521f : SUCCESS in 3m 27s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/0230af761ace4b508dd6df07897727c1 : SUCCESS in 51m 47s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/fb54a5920a2648d48a8b7ae7f930a9fd : FAILURE in 51m 48s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/94ec9a7ae2a2408895a2927ee09fb085 : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":7},{"id":"eff5dbf82d1b688d5a7a75564d7cfca7cb19bb7e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-23 00:01:29.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"18d79e34a0df641923a49da5a83d75179fb573d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 00:55:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7de5798276d4928a3082b66d1b0c4c5 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80eef367bd334d3e9c0b5354ab74d1ea : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d44c03fbf0f46529dc182e2c7483dfa : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cde9a0a991d4219be58beadc35f287e : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a888cb3eabbb4691932b88dc567527a1 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f83fddc23d4539a3734fa3d677e5a3 : SUCCESS in 3m 22s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/dde3acdfc74646c19f9f017ff1d345e9 : SUCCESS in 44m 30s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/33ca84294add4c2e9d31b1e1a31be403 : SUCCESS in 44m 04s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/dde2ebbf4bad40eba5d4ef0e48d4d633 : SUCCESS in 48m 06s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/d524f07362004586b4dd3dab232d071d : SUCCESS in 37m 39s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/233e430cb1944c2ab1d966349e27a00d : SUCCESS in 38m 20s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/fdd8468718fb4e92bd909a0c915f982a : FAILURE in 46m 15s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/a25e2ac3068545dbb3d634245c01cfb8 : FAILURE in 46m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e09090703c8b444e96efe2645e283c93 : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a8c265d796d2e1b0a12467cbb0184d7be0b067b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 00:56:14.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c94bbbf63e95f010b0ed2461296eefa1c653b329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 01:49:22.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e68b4a21a3ff4e84a920349b6890e737 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e616136a7cd43f4896f939e13461be6 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c03e65c2b9a64e4ea9649bda958a1d78 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/428271b072dd4be9b6d668e8ccbbe932 : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/844c711de8c64a438f7ed0cffe9a4b6b : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc5120805f44344a21897dc4e916d61 : SUCCESS in 6m 06s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/6858184d4a8f4aa98bdf1d60cbcf4569 : SUCCESS in 50m 40s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/056502e54b9c4c07a399fab3f50eb0ae : SUCCESS in 50m 53s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/6db64bd18a0b487ead1a0c7b3e729ccb : SUCCESS in 46m 15s","accounts_in_message":[],"_revision_number":7},{"id":"f15d1711086bf898f11193de8093b441736a13bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 01:49:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"b39555ea2ec9a130ec50ee376dce4dd5629b3d51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 01:51:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f493712ac894b0cbc05016da8540e5e : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b860edd8385d551eb0a886ab6706425fce605d79","revisions":{"df7675d3e4dbeb291ae3f33faa07936a5f0a9629":{"kind":"REWORK","_number":1,"created":"2019-08-16 11:32:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/676910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/10/676910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/10/676910/1"}}},"commit":{"parents":[{"commit":"70645ff72768cd639a2182e8e2ba8aff9605f9b7","subject":"Update hacking version to latest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/70645ff72768cd639a2182e8e2ba8aff9605f9b7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-16 11:29:52.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-16 11:29:52.000000000","tz":0},"subject":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate","message":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027senlin-tempest-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027senlin-tempest-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35929\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ib7695212e42ef16dbc1edd65c3516b277e99078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/df7675d3e4dbeb291ae3f33faa07936a5f0a9629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/df7675d3e4dbeb291ae3f33faa07936a5f0a9629"}]},"branch":"refs/heads/master"},"44ae31fa53524ec8b491f4da3fb3ef2e509d0845":{"kind":"REWORK","_number":2,"created":"2019-08-16 18:07:13.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/676910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/10/676910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/10/676910/2"}}},"commit":{"parents":[{"commit":"70645ff72768cd639a2182e8e2ba8aff9605f9b7","subject":"Update hacking version to latest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/70645ff72768cd639a2182e8e2ba8aff9605f9b7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-16 11:29:52.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-16 18:07:10.000000000","tz":0},"subject":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate","message":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027senlin-tempest-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027senlin-tempest-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35929\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ib7695212e42ef16dbc1edd65c3516b277e99078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/44ae31fa53524ec8b491f4da3fb3ef2e509d0845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/44ae31fa53524ec8b491f4da3fb3ef2e509d0845"}]},"branch":"refs/heads/master"},"266c05b9c21b178670711020fbacb974cda25315":{"kind":"REWORK","_number":3,"created":"2019-08-18 14:15:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/676910/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/10/676910/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/10/676910/3"}}},"commit":{"parents":[{"commit":"70645ff72768cd639a2182e8e2ba8aff9605f9b7","subject":"Update hacking version to latest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/70645ff72768cd639a2182e8e2ba8aff9605f9b7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-16 11:29:52.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-18 14:15:36.000000000","tz":0},"subject":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate","message":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027senlin-tempest-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027senlin-tempest-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35929\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ib7695212e42ef16dbc1edd65c3516b277e99078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/266c05b9c21b178670711020fbacb974cda25315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/266c05b9c21b178670711020fbacb974cda25315"}]},"branch":"refs/heads/master"},"e4960695dc65d0b4a4529ba9e7155de74ec1088a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-10-18 05:55:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/676910/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/10/676910/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/10/676910/4"}}},"commit":{"parents":[{"commit":"70645ff72768cd639a2182e8e2ba8aff9605f9b7","subject":"Update hacking version to latest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/70645ff72768cd639a2182e8e2ba8aff9605f9b7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-16 11:29:52.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-18 05:55:30.000000000","tz":0},"subject":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate","message":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027senlin-tempest-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027senlin-tempest-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35929\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nDepends-On: https://review.opendev.org/#/c/688253/\nChange-Id: Ib7695212e42ef16dbc1edd65c3516b277e99078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e4960695dc65d0b4a4529ba9e7155de74ec1088a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e4960695dc65d0b4a4529ba9e7155de74ec1088a"}]},"branch":"refs/heads/master"},"00f0a35d56294320ff9d17ee4c0a04f759a7a24f":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-10-18 05:56:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/676910/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/10/676910/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/10/676910/5"}}},"commit":{"parents":[{"commit":"70645ff72768cd639a2182e8e2ba8aff9605f9b7","subject":"Update hacking version to latest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/70645ff72768cd639a2182e8e2ba8aff9605f9b7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-16 11:29:52.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-18 05:56:05.000000000","tz":0},"subject":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate","message":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027senlin-tempest-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027senlin-tempest-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35929\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ib7695212e42ef16dbc1edd65c3516b277e99078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/00f0a35d56294320ff9d17ee4c0a04f759a7a24f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/00f0a35d56294320ff9d17ee4c0a04f759a7a24f"}]},"branch":"refs/heads/master"},"459d1993d26342d1d53b4d208442b254d77ebdc2":{"kind":"REWORK","_number":6,"created":"2019-10-21 18:43:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/676910/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/10/676910/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/10/676910/6"}}},"commit":{"parents":[{"commit":"70645ff72768cd639a2182e8e2ba8aff9605f9b7","subject":"Update hacking version to latest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/70645ff72768cd639a2182e8e2ba8aff9605f9b7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-16 11:29:52.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-21 18:43:22.000000000","tz":-420},"subject":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate","message":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027senlin-tempest-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027senlin-tempest-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35929\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ib7695212e42ef16dbc1edd65c3516b277e99078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/459d1993d26342d1d53b4d208442b254d77ebdc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/459d1993d26342d1d53b4d208442b254d77ebdc2"}]},"branch":"refs/heads/master"},"b860edd8385d551eb0a886ab6706425fce605d79":{"kind":"REWORK","_number":7,"created":"2019-10-21 19:59:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/676910/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/10/676910/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/10/676910/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/10/676910/7"}}},"commit":{"parents":[{"commit":"70645ff72768cd639a2182e8e2ba8aff9605f9b7","subject":"Update hacking version to latest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/70645ff72768cd639a2182e8e2ba8aff9605f9b7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-16 11:29:52.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-21 19:59:18.000000000","tz":-420},"subject":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate","message":"[train][goal] Define new \u0027senlin-tempest-api-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit adds the new job \u0027senlin-tempest-api-ipv6-only\u0027\nrun on gate which is derived from \u0027devstack-tempest-ipv6\u0027.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen address\n- \u0027senlin-tempest-api-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35929\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ib7695212e42ef16dbc1edd65c3516b277e99078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/b860edd8385d551eb0a886ab6706425fce605d79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/b860edd8385d551eb0a886ab6706425fce605d79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
