)]}'
{"id":"openstack%2Fsenlin~691504","triplet_id":"openstack%2Fsenlin~master~I0858af84509356326323fd224498256ff585a150","project":"openstack/senlin","branch":"master","topic":"dup_code","hashtags":[],"change_id":"I0858af84509356326323fd224498256ff585a150","subject":"Removed duplicate notification implementations","status":"MERGED","created":"2019-10-27 06:22:47.000000000","updated":"2019-11-27 03:26:44.000000000","submitted":"2019-11-27 03:25:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691504-1574825127683-2b47506d","meta_rev_id":"c02889cd0d6570d28ce7b2afbb9bd893e1e32c53","_number":691504,"virtual_id_number":691504,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 03:25:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":2,"date":"2019-11-27 02:32:20.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":2,"date":"2019-11-21 20:50:02.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":1,"date":"2019-11-27 02:32:20.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 19:46:53.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2019-11-27 02:32:20.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2019-11-27 03:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a06cc8cf23733cd77d9bca1faaa02aa5588b6a80","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-27 06:22:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46703e4b481a2ec3afb9345b551973a5b96fd26c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-27 06:36:42.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"66124ba27272d94864f08ac5523706dfbe520160","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 07:38:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/950896779cbc46cea186b1f4b1e72c56 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a236a80326d440cdae6c0b2d8a9d4767 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47f182035a064b5f85bcf4d8bc082f26 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8811764990e4b1cbb1c9112d554cc3a : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13b39ae994494929bec6c72417e35b90 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb122d48818642a0956ed261660409be : SUCCESS in 4m 30s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/f6ea7b04c4b6490dbc9b1d4b5b74db9a : SUCCESS in 50m 01s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/b956801e81bd458db82bee2f962a242f : SUCCESS in 42m 23s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/5da054febca34c56a18f665ddb575051 : SUCCESS in 55m 25s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/ea2898bfe3444b0886ff78badd2e3d4c : SUCCESS in 38m 29s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/0bf1c5b8fdf94e20a892a72300d1024a : SUCCESS in 47m 31s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/4b925587ad2a46a6aa4eab86f9742916 : FAILURE in 54m 26s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/f3ba8e880d0a44feba5e2994280f6411 : FAILURE in 44m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd7413d37e88470d95ff63bb33c17224 : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"da17f03a3e90493b559ea7e4c0efc7da5a96cd29","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-28 02:25:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1421d8984e93b58c22325b5d7dec85c0b3d35acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 03:27:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04e19486b5384af2b4e1ad0975236cb5 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c416f2e0a3084fd697f71aeabf8350aa : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ce66f0fd36442068c6507fc63e02157 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40d778cb0d064549922e5b6a75e449d4 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94451f68928145cb9f1f59e4136c4f5c : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36f32d4159c14006b7b7074937ca6977 : SUCCESS in 4m 03s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/b6e18aaf827b4d6aaa4941bff6e1c40d : SUCCESS in 59m 10s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/2846d92a7e41473f9c5db4966b15aba6 : SUCCESS in 58m 54s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/1d8dfac2ca964a00a1f178fe5e370723 : SUCCESS in 42m 39s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/54a214d617de4c5a9f89c09358caa618 : SUCCESS in 40m 10s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/a3d2f423285b4c5398efce5effaefb60 : SUCCESS in 54m 40s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/0a785c74a1c44f66b5b8f83cc5cac010 : FAILURE in 41m 46s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/d376dc7c400f40419bb0851067828ec2 : FAILURE in 58m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/308f82427dcb4b1394626d6ef15096ea : SUCCESS in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ff6b487a9792c129b7066b7fb20af340c942143","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-28 03:43:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"265ad4fea28137902fd7166aeb775389111fa0d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 04:28:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8df707c67ca4699b4dba1fdac336345 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ca6c8aa98294264a898ba7012276699 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd36c73b64aa48f2859ee62edfde80d4 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc7046104a204389a134631587ffbd50 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03677c0fe6e4443dbe231b9023962a26 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78117f94aa2d427084cd2c37cde1e402 : SUCCESS in 4m 03s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/7a766bac2ed3429b9551e565e9b8b480 : SUCCESS in 38m 14s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/f430b2ae0fb948a6a3e9ae6a2e1f9ec4 : SUCCESS in 35m 17s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/3fe72aa4cb57477e9cef4ad942231d36 : SUCCESS in 43m 08s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/6eab0b9408bf48879fc78abbe982b3f9 : SUCCESS in 34m 58s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/f04ad7d8749e4430ac5b848eb47231b2 : SUCCESS in 37m 52s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/ef6fe4fecfa8486db2a7b97d7b0b206a : FAILURE in 40m 53s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/0618d9c535c44e619cbcb9479415c71d : FAILURE in 40m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f4d92988415439296138e4274f9ebcd : SUCCESS in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b415b9c35742757b3378c2cb8eac0943667bba0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-28 05:00:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7caf3d2d1d45f13c5807d49c1259c41424b195f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 05:45:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcfbdbe05a2e40af8a24eefe7826ea14 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8edf53be3214f39ac51ed043fd67145 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57db74048163486dbb1088236d7e8355 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fec0a1860bc49cfacc810ffc63d8bb2 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dae470fe97f842bb8254aec9361be152 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86d8f272cec48cca5562abc566e2217 : SUCCESS in 3m 51s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/1c81e3c540634fb091897ded1b912b65 : SUCCESS in 44m 16s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/46baded5e40b4cdfb81aa3ae13ab42be : SUCCESS in 35m 44s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/b2f7486d7198469cbb86ab0e3059442b : SUCCESS in 43m 02s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/7ae8119174aa488e915f844cc45fd2bb : SUCCESS in 32m 39s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/e78d3d805e2a49b4a9721e56e1b5c430 : SUCCESS in 36m 57s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/8943bc2ee9c04409a3a6914c0e850113 : FAILURE in 41m 41s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/3295ffdaa32541cc982a3e00486ef2f5 : FAILURE in 42m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50b2fbcf3c7f4ed0b4a6c6b4a436942b : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be07685c6e08e478f02d9c40b50de60ab9692b15","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-10-28 16:36:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"704a3e2721c5265159022bf3a2147e1423514cdc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-28 17:12:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9f6db5e5b4bc6bd19ea8fc8aa7f876d135a0abc1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-28 17:49:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"339ba16dc05fe69ef41c2df8d41c81abe04c7b9f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-28 17:59:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"685605d8643149b812d18fb6b3da98efe6925fa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:46:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d0d9b3c49fb4bb6859810e6924fa21e : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/667bca3eaa8e4e31852c14ac32602133 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1a2baa4d7234d56a4008524e74e5407 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f0254693be3462699df92ac191d24d8 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cc7908524ae4cbf98253596045c44ed : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6960058a187547b1adf1071275c23e6a : SUCCESS in 4m 04s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/351c3a81e2ca4cc7bd299dc33afe4ead : SUCCESS in 42m 40s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/07e2ecc7eaf34f7892d21e897b31babc : SUCCESS in 42m 29s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/6c9ac4571dde464e92a578a5b4762e80 : SUCCESS in 43m 19s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/3ed2ffc554a04375a8768ac319319ad9 : SUCCESS in 38m 47s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/c04581eee7ec4fb4878f2e7714f22ba3 : SUCCESS in 39m 46s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/74d7fba1177545a08f5bbe913aac3a79 : FAILURE in 40m 33s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/65b5b84560114940a2a41d9344492d5e : FAILURE in 40m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17eb74e3f255447b86c664ab4d25d43c : SUCCESS in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"652ee91662c736c861b76025ce8c3fa6e8a8e44a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-30 08:09:17.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"e1ebd3634973169385f4741270acf20f44d97d63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 09:12:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/599a8807fffb4ab08585c08b6976eb63 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16404c29dce40d98e10268a490d0bc4 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32a0c87d10e04e4397de39c3f21f2683 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5ec2254ff8e4c8c99b1fd76dfd50373 : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0fe85b1e80541d68801371f02a74ec8 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b0e0d1a4dd54921828c8e0e39676322 : SUCCESS in 4m 27s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/893a6a1f60474f659afd73ccef2653d6 : SUCCESS in 45m 12s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/cd095f5c6e7a4d45996eb4fbccfce57c : SUCCESS in 50m 48s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/4e3ec040e95945218fd0e1caf221908c : SUCCESS in 44m 57s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/70fd9c37d3f448e0b1b0a93591b656ab : SUCCESS in 41m 49s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/07484674b89b4de087acb928d42c88f7 : SUCCESS in 39m 59s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/8e53131ee600455996636621c8004700 : FAILURE in 45m 14s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/5a6ca4a10187460d8e8246f4e1204e67 : FAILURE in 44m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54538be8581b438f9d78324c4f591c2b : SUCCESS in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"16599d18d7a3e287872ec5f294d1e755c7002d23","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-14 22:40:48.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"b2a90a91ef3cda6fcc032b49038f5a83cc04fe13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 23:31:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb5cdb37d7654737926cd7394c322501 : FAILURE in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1927344c0dc4f87a05486eb7893b79e : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95b3e6f22382486496e7c3120f0e073c : FAILURE in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a01ad2593b10490088e1561017922a67 : FAILURE in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c8ea3ee13794b01a3d88decb279a91f : FAILURE in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e91fbc17aa0459287156cb138d97659 : SUCCESS in 4m 02s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/c6c6e83715f74f8da2ddb132e678cce5 : SUCCESS in 42m 31s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/ce4d6a8f356b49dbb5efb41429a056dd : SUCCESS in 44m 29s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/685934ceb4aa4e36ade2b5473ed286cb : SUCCESS in 42m 49s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/7653651ecee34d458967044525b71bdc : SUCCESS in 41m 12s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/81f06810d51d4f849b3e372402bf844f : SUCCESS in 37m 34s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/e885b4cb021643169f9b0ccb51c1bc91 : FAILURE in 40m 24s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/fa2446cd383e4bfa81ffa3a125320e6a : FAILURE in 45m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18a6d82e5d704b218c9b5d98fed9683d : FAILURE in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d5b17853790e8d231d80e6d874335a2c4d50c88b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-15 06:41:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"140f42c6511f907e755001eed9ba414dd2fc130e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 07:46:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d45c3150109241d6bcba2837de6450fe : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e62bb4d9580a4d92ad5760b46f1025c4 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66d0fcf221d34b129386330a9dc2250a : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/461fe36ea2fe4908ac1cb7dc053a65fd : SUCCESS in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bab2c90cfdbf496d8396d83c01b50fc8 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0a2851faf9a4ef99d879eed67c018b4 : SUCCESS in 7m 04s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/7390272dc8284900812a68555962fad8 : SUCCESS in 50m 41s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/aedfee2b04834f7281845b2bbbce228f : SUCCESS in 46m 50s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8440b8c367eb41c7bb75e872b292717c : SUCCESS in 55m 09s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/3030d3903fa542fe9637d5c6a25b30d4 : SUCCESS in 40m 16s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/cadd7713704049248ae9c3654387ef4a : SUCCESS in 45m 14s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/1f21434380754f7f9c66b7d21612c695 : FAILURE in 54m 06s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/190d2770a5674e87a23543af980dff6f : FAILURE in 54m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c75eba6326ae4c3bb6f0a9ebe4d96306 : SUCCESS in 7m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ded08d681f291fdeead4c3d43a65e74e9b5fdc95","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-11-15 17:13:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b4fe5c1b206db3761cfea6a4138b53a73d0e3683","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-15 18:14:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d8042a6ffce697159bc6b0b31a72f739a3ba496c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-15 18:43:21.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"c2fba49eb715e0a05a3ea037163b96b29a688e5d","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-11-15 19:46:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"098d6ff1d5557397dfdee6a0534ed190ddf854b3","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-11-15 19:46:53.000000000","message":"Patch Set 10: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":10},{"id":"357498f7f7aa892c3bbb56dacd4e582f7aff3f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 19:50:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5007211e7b424be49e8440e6795a12c7 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8460f511e2747079411d2716539a951 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b31e1913dcdb41bd84a4e34ee1fb3199 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65f7ddee8e3a4130a4889bc0a57d8a47 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b4792729c6a45f4b3f986218f59a7c3 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9625e3c4e7c431ca548bf2864a53d8a : SUCCESS in 4m 03s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/e70dcaffedb04e498c4f1590cc9860d5 : SUCCESS in 45m 57s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/70404042d4454a85addd462dbdf0cd42 : SUCCESS in 39m 38s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8fb8d19ccefd40e1ae20ad4fe0aef00c : SUCCESS in 43m 45s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/9df3139c83ef40c68f0f4bb504602f58 : SUCCESS in 34m 33s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/106cf790813543c9938bd322904453b7 : SUCCESS in 39m 56s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/dcc6864b0acb4c109eb200594788cea5 : FAILURE in 41m 21s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/c79e686b5cff4e0bab5a6e059cb5da12 : FAILURE in 59m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fe65096df3c4067a246558475463cfe : SUCCESS in 6m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8eb05722fcb43e7a830205349382cf92c3c36ca8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-21 20:50:02.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"2449e855a03c9fb11248b4971ca438c364547657","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 21:33:35.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8812a9dc705d460ba74ec3278b9bae5c : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33e81234c4cd4d21932819832e465fe5 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41cbf5a70e9f4b83b037b317bff02c60 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31890802f5d34a7fa4cc03fb34bdf887 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d38aa4e478d9412b8a414a20de418835 : SUCCESS in 4m 00s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/8b784ef89b334e1c9110baa220c038f3 : SUCCESS in 40m 15s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/000f1ce2826743e885a4f48be495d818 : SUCCESS in 41m 54s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/cf315968da6e4dc8b606061e9ab736ef : SUCCESS in 38m 45s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/ec223fe29eed44d3b0f27faf05e5232a : FAILURE in 40m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69a8f5fcea3d4e0f8912058e39fb8b7d : SUCCESS in 6m 40s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"60898aae551e338e25aece36dc59bb2dd792b4e2","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-11-27 02:32:20.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":11},{"id":"019048f2ee1bfa41657a6415f59babbfc5587bf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 02:32:31.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"578b275cfeba1af586cee3afd6bf4ab57f84a49a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 03:25:27.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdcf09fb37704986bd64bdb6926b3254 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e480666b97b4267b14335ae1fc5f8fe : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9b9f44b10ed460787781c8e415c19cc : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3be6f31bc068498bab01a001159d2f08 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497dfed08e924f6e97bdb187b319528c : SUCCESS in 4m 22s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/a498b7dfb2154d48a5e1a34024b45a6b : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":11},{"id":"aea053f4b3cc4b85b41ae590846a931cb7e03ac4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 03:25:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"e189d35675dc6f11812a45d63a2bea67f5f98162","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 03:26:44.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a624c813ff764793b13135401b6930ac : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"3d20220f34cab9f958063b556ff6aa1f5573ad19","revisions":{"e0ec4580596922c89483b2ccdffc9a4e65b7788f":{"kind":"REWORK","_number":1,"created":"2019-10-27 06:22:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/691504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/04/691504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/04/691504/1"}}},"commit":{"parents":[{"commit":"ac493538991b5ea0a06bf7904efe9ebc4bf7333b","subject":"Merge \"Add tainted field to nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ac493538991b5ea0a06bf7904efe9ebc4bf7333b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"subject":"Removed duplicate notifactions","message":"Removed duplicate notifactions\n\nWe somehow ended up with duplicate implementations\nfor nova and heat endpoints.\n\nChange-Id: I0858af84509356326323fd224498256ff585a150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e0ec4580596922c89483b2ccdffc9a4e65b7788f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e0ec4580596922c89483b2ccdffc9a4e65b7788f"}]},"branch":"refs/heads/master"},"ef999a5ac9974a4044d15bc2cf53d8e2d7c71c9d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-27 06:36:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/691504/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/04/691504/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/04/691504/2"}}},"commit":{"parents":[{"commit":"ac493538991b5ea0a06bf7904efe9ebc4bf7333b","subject":"Merge \"Add tainted field to nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ac493538991b5ea0a06bf7904efe9ebc4bf7333b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:36:40.000000000","tz":0},"subject":"Removed duplicate notifications","message":"Removed duplicate notifications\n\nWe ended up with duplicate implementations\nfor nova and heat endpoints. This patch\nremoves the duplicates and merges the code.\n\nChange-Id: I0858af84509356326323fd224498256ff585a150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ef999a5ac9974a4044d15bc2cf53d8e2d7c71c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ef999a5ac9974a4044d15bc2cf53d8e2d7c71c9d"}]},"branch":"refs/heads/master"},"2cd2b01d6278fb4d33f166272d4acca4e28f55bf":{"kind":"REWORK","_number":3,"created":"2019-10-28 17:12:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/691504/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/04/691504/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/04/691504/3"}}},"commit":{"parents":[{"commit":"ac493538991b5ea0a06bf7904efe9ebc4bf7333b","subject":"Merge \"Add tainted field to nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ac493538991b5ea0a06bf7904efe9ebc4bf7333b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-28 17:12:41.000000000","tz":-420},"subject":"Removed duplicate notifications","message":"Removed duplicate notifications\n\nWe ended up with duplicate implementations\nfor nova and heat endpoints. This patch\nremoves the duplicates and merges the code.\n\nChange-Id: I0858af84509356326323fd224498256ff585a150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/2cd2b01d6278fb4d33f166272d4acca4e28f55bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/2cd2b01d6278fb4d33f166272d4acca4e28f55bf"}]},"branch":"refs/heads/master"},"38dd80e3550b9547a9dbabeb4574c5c173ec7feb":{"kind":"REWORK","_number":4,"created":"2019-10-28 17:49:15.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/691504/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/04/691504/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/04/691504/4"}}},"commit":{"parents":[{"commit":"ac493538991b5ea0a06bf7904efe9ebc4bf7333b","subject":"Merge \"Add tainted field to nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ac493538991b5ea0a06bf7904efe9ebc4bf7333b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-28 17:49:27.000000000","tz":-420},"subject":"Removed duplicate notifications","message":"Removed duplicate notifications\n\nWe ended up with duplicate implementations\nfor nova and heat endpoints. This patch\nremoves the duplicates and merges the code.\n\nChange-Id: I0858af84509356326323fd224498256ff585a150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/38dd80e3550b9547a9dbabeb4574c5c173ec7feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/38dd80e3550b9547a9dbabeb4574c5c173ec7feb"}]},"branch":"refs/heads/master"},"07f2f08312a58f2466ff08dd95bf57613dd76929":{"kind":"REWORK","_number":5,"created":"2019-10-28 17:59:48.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/691504/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/04/691504/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/04/691504/5"}}},"commit":{"parents":[{"commit":"ac493538991b5ea0a06bf7904efe9ebc4bf7333b","subject":"Merge \"Add tainted field to nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ac493538991b5ea0a06bf7904efe9ebc4bf7333b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-28 17:59:57.000000000","tz":-420},"subject":"Removed duplicate notifications","message":"Removed duplicate notifications\n\nWe ended up with duplicate implementations\nfor nova and heat endpoints. This patch\nremoves the duplicates and merges the code.\n\nChange-Id: I0858af84509356326323fd224498256ff585a150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/07f2f08312a58f2466ff08dd95bf57613dd76929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/07f2f08312a58f2466ff08dd95bf57613dd76929"}]},"branch":"refs/heads/master"},"37d853dde0a186a55612e818107cbba93507dc4f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-10-30 08:09:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/691504/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/04/691504/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/04/691504/6"}}},"commit":{"parents":[{"commit":"2beb072ec246a002f356e7b939f81de4bebbce07","subject":"Merge \"Enable Apache by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/2beb072ec246a002f356e7b939f81de4bebbce07"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-30 08:09:17.000000000","tz":0},"subject":"Removed duplicate notifications","message":"Removed duplicate notifications\n\nWe ended up with duplicate implementations\nfor nova and heat endpoints. This patch\nremoves the duplicates and merges the code.\n\nChange-Id: I0858af84509356326323fd224498256ff585a150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/37d853dde0a186a55612e818107cbba93507dc4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/37d853dde0a186a55612e818107cbba93507dc4f"}]},"branch":"refs/heads/master"},"7ccacb2e12c5d622c3be2eb09751de562a10fdb2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-11-14 22:40:48.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/691504/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/04/691504/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/04/691504/7"}}},"commit":{"parents":[{"commit":"fdd714dd5a715901d36e975dbf524a62c556a04f","subject":"Merge \"Bumping worker count during testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/fdd714dd5a715901d36e975dbf524a62c556a04f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-14 22:40:48.000000000","tz":0},"subject":"Removed duplicate notifications","message":"Removed duplicate notifications\n\nWe ended up with duplicate implementations\nfor nova and heat endpoints. This patch\nremoves the duplicates and merges the code.\n\nChange-Id: I0858af84509356326323fd224498256ff585a150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7ccacb2e12c5d622c3be2eb09751de562a10fdb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7ccacb2e12c5d622c3be2eb09751de562a10fdb2"}]},"branch":"refs/heads/master"},"e509b1d193a0069fd3229afd24c7a71b71e122db":{"kind":"REWORK","_number":8,"created":"2019-11-15 06:41:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/691504/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/04/691504/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/04/691504/8"}}},"commit":{"parents":[{"commit":"fdd714dd5a715901d36e975dbf524a62c556a04f","subject":"Merge \"Bumping worker count during testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/fdd714dd5a715901d36e975dbf524a62c556a04f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-15 06:41:29.000000000","tz":-480},"subject":"Removed duplicate notifications","message":"Removed duplicate notifications\n\nWe ended up with duplicate implementations\nfor nova and heat endpoints. This patch\nremoves the duplicates and merges the code.\n\nChange-Id: I0858af84509356326323fd224498256ff585a150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e509b1d193a0069fd3229afd24c7a71b71e122db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e509b1d193a0069fd3229afd24c7a71b71e122db"}]},"branch":"refs/heads/master"},"09b7b6bae1373c7cb1e927f4650b691181408a48":{"kind":"REWORK","_number":9,"created":"2019-11-15 18:14:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/691504/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/04/691504/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/04/691504/9"}}},"commit":{"parents":[{"commit":"fdd714dd5a715901d36e975dbf524a62c556a04f","subject":"Merge \"Bumping worker count during testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/fdd714dd5a715901d36e975dbf524a62c556a04f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-15 18:14:06.000000000","tz":-480},"subject":"Removed duplicate notifications","message":"Removed duplicate notifications\n\nWe ended up with duplicate implementations\nfor nova and heat endpoints. This patch\nremoves the duplicates and merges the code.\n\nChange-Id: I0858af84509356326323fd224498256ff585a150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/09b7b6bae1373c7cb1e927f4650b691181408a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/09b7b6bae1373c7cb1e927f4650b691181408a48"}]},"branch":"refs/heads/master"},"0d8aad025db9aa16e6d87af97788b63cced1d077":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2019-11-15 18:43:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/691504/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/04/691504/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/04/691504/10"}}},"commit":{"parents":[{"commit":"fdd714dd5a715901d36e975dbf524a62c556a04f","subject":"Merge \"Bumping worker count during testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/fdd714dd5a715901d36e975dbf524a62c556a04f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-15 18:43:19.000000000","tz":0},"subject":"Removed duplicate notification implementations","message":"Removed duplicate notification implementations\n\nWe ended up with duplicate implementations\nfor nova and heat endpoints. This patch\nremoves the duplicates and merges the code.\n\nChange-Id: I0858af84509356326323fd224498256ff585a150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/0d8aad025db9aa16e6d87af97788b63cced1d077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/0d8aad025db9aa16e6d87af97788b63cced1d077"}]},"branch":"refs/heads/master"},"3d20220f34cab9f958063b556ff6aa1f5573ad19":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-11-21 20:50:02.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/691504/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/04/691504/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/04/691504/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/04/691504/11"}}},"commit":{"parents":[{"commit":"29c0e6fbdfb8e91d0c2bfc6e4fe3f0c317b5504f","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/29c0e6fbdfb8e91d0c2bfc6e4fe3f0c317b5504f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-27 06:22:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-21 20:50:02.000000000","tz":0},"subject":"Removed duplicate notification implementations","message":"Removed duplicate notification implementations\n\nWe ended up with duplicate implementations\nfor nova and heat endpoints. This patch\nremoves the duplicates and merges the code.\n\nChange-Id: I0858af84509356326323fd224498256ff585a150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/3d20220f34cab9f958063b556ff6aa1f5573ad19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/3d20220f34cab9f958063b556ff6aa1f5573ad19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
