)]}'
{"id":"openstack%2Fsenlin~691589","triplet_id":"openstack%2Fsenlin~master~I6ab0a8237af1431baa75e7aaf09cabab526e53fd","project":"openstack/senlin","branch":"master","hashtags":[],"change_id":"I6ab0a8237af1431baa75e7aaf09cabab526e53fd","subject":"Stop testing python2.7","status":"ABANDONED","created":"2019-10-28 05:54:44.000000000","updated":"2019-11-29 21:05:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1282055b24930e478076ccf9de299508cea6d6e0","_number":691589,"virtual_id_number":691589,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-28 06:45:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2019-10-28 16:58:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2019-11-29 21:05:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 06:45:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 16:58:16.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-11-29 21:05:38.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"d1aa6af1c1b8e143c9e5bad3d66b06e22f630c78","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-10-28 05:54:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ad3cb0415241acd50aaca9230bba988d4d61ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 06:45:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9c0cb11a96424affae7b6c6042e2bb33 : SUCCESS in 3m 52s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4b25dba8ff204f53990c7ce214498e90 : SUCCESS in 2m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e72076602b3749fc9ea9c7a58d8f7bc8 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b16fd997082047cf9c7c0c02ea8b0a1a : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eae126a4864a4fb092e704c36e787429 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcf7b302c4494327bb489d2d6da69441 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d7dcadd26d748e498940c0930ca6b23 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9a2183ebd64efb89e2215b1322de70 : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09d89db712de4ea5abbeb7937cae7758 : SUCCESS in 9m 25s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/f1544a1e23994a819c900dffb9571701 : SUCCESS in 45m 30s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/b57904192d3c40018d0c91632ddaac50 : SUCCESS in 42m 00s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/883b7fe438454213a5e45e8133ada4dc : SUCCESS in 43m 35s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/2e17ce47b1654877a9bb003102b38958 : SUCCESS in 37m 59s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/3554f25973214bb4bc4428eaf04550b3 : SUCCESS in 36m 13s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/9d4fe1a67f454379925d917c18a33819 : FAILURE in 46m 41s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/a8176f0be9d24d2e835cd552b946ef0d : FAILURE in 40m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad7ef32e450a42cfa5cd2529cfcc0837 : SUCCESS in 6m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79bb99e557c24af9a93f9078d9161b1d519aaea3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-28 16:58:16.000000000","message":"Patch Set 1: Code-Review-1\n\nwe need to do a lot more on this. I am composing the community-wide goal  -https://review.opendev.org/#/c/691178/\n\nLet\u0027s be on sync on what all to do and when via goal.","accounts_in_message":[],"_revision_number":1},{"id":"5fe9b51e2d73ca4cb9f8f25e4cea5d5daf0361ca","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-29 21:05:38.000000000","message":"Patch Set 1: Code-Review-2\n\nThis is already done here.\nhttps://review.opendev.org/#/c/694701/","accounts_in_message":[],"_revision_number":1},{"id":"1282055b24930e478076ccf9de299508cea6d6e0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-29 21:05:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ada43d43e8a6cc2133563fdb6428e9fd4939658","revisions":{"4ada43d43e8a6cc2133563fdb6428e9fd4939658":{"kind":"REWORK","_number":1,"created":"2019-10-28 05:54:44.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/89/691589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/89/691589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/89/691589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/89/691589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/89/691589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/89/691589/1"}}},"commit":{"parents":[{"commit":"6d566c9a4202d49a3f955bca2ccd7bf9d8e53d9f","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/6d566c9a4202d49a3f955bca2ccd7bf9d8e53d9f"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-28 05:53:54.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-28 05:53:54.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nChange-Id: I6ab0a8237af1431baa75e7aaf09cabab526e53fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/4ada43d43e8a6cc2133563fdb6428e9fd4939658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/4ada43d43e8a6cc2133563fdb6428e9fd4939658"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
