)]}'
{"id":"openstack%2Fsenlin~692691","triplet_id":"openstack%2Fsenlin~master~Ieaec56e66e6804c09827f39bf68d9ec935c42183","project":"openstack/senlin","branch":"master","topic":"new_opts","hashtags":[],"change_id":"Ieaec56e66e6804c09827f39bf68d9ec935c42183","subject":"Standardize worker / thread config","status":"MERGED","created":"2019-11-03 01:56:31.000000000","updated":"2019-11-14 21:28:45.000000000","submitted":"2019-11-14 21:27:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692691-1573766834019-a4121484","meta_rev_id":"b41cb1465341e795b12f892531e906cfc2501cc2","_number":692691,"virtual_id_number":692691,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 21:27:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"recommended":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":1,"date":"2019-11-05 02:01:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":2,"date":"2019-11-05 02:27:42.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":2,"date":"2019-11-04 23:48:14.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":1,"date":"2019-11-05 02:27:42.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 23:48:14.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2019-11-05 02:01:34.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-11-05 02:27:42.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2019-11-14 21:27:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e08400e35c7d0643d97078de1c99e3cce54ffe43","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-03 01:56:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5a57ec1b5ba46d198d2157d290cc287d21eb8ab","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-03 02:01:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"806ee8a58f25b24106ac9d42f7307a9f414153b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 02:50:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df1a70ea9d764861a00c597502a964db : FAILURE in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7cd82224837477d8851bd5f4cbd4d19 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5396847ce68043c1b83937c8a1b45bf6 : FAILURE in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c271d0b1a7564ed29c1d334c8d846812 : FAILURE in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e861675ce1e4b8fad2213da78d66b91 : FAILURE in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54d6cde0e304f79969aefc3975017db : SUCCESS in 4m 15s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/1e6a3d720a83409da2e83bb4209d7d3e : SUCCESS in 42m 45s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/d73791171f6b41aa96bb0fda58fd5e62 : SUCCESS in 42m 35s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/931dcd581e9147ec8f0ce2aefb184f12 : SUCCESS in 40m 22s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/eb3d6c65c9f74434b9864c9416f2e3da : SUCCESS in 37m 52s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/822d284bff0a4a298de5cdcdde93132c : SUCCESS in 39m 36s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/3ee291bcad304da5ae9958b1a2259770 : FAILURE in 42m 56s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/32f5503e1f04444ba02201ceed4ef317 : FAILURE in 43m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fce22ce3560541ceaf70186e57c13688 : FAILURE in 6m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77590627195e90f136419731778c7e2ed163af6b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-03 02:53:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0e0e6c40bba784656ac860b25bc24300c665ed29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 03:40:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7eda122469fd46f2bb039ba2764ab1e3 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beef3d5c7b3640beb60a5533bfecdf03 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7163c29917a4081907d431e14d88f5a : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/730374dbf5234d348ba133b0a36e81ad : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68204df87a0747449fded98bcd750f91 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3850d7df524ccd9d8aedcd81354b98 : SUCCESS in 4m 05s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/12387acbdc0d4498a416b9f3bdf23dff : SUCCESS in 41m 04s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/e01d07f989194b1795e8c25d0d034774 : SUCCESS in 42m 40s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/292466bd8ab84e26a643000e1326f815 : SUCCESS in 40m 59s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/3bad4de70bd542dcb5b0caed3a4f6170 : SUCCESS in 36m 43s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/7becb564142542e685583638e9b4ecf0 : SUCCESS in 39m 54s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/332f2073e4fc4d6fa5c0bf5d8b8e6779 : FAILURE in 42m 34s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/25825f2d3fc84df2bd959df3f6e87957 : FAILURE in 44m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbc030c845d2448080672c0e61cac77d : FAILURE in 5m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f2af2e5dde704d3da502de3db8c7910d49d3fa9f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-03 09:00:19.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"117bfa341e62c02b0f6586a738243c80c2f503da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 10:06:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a417f1844e648be9d82f1a79eef50a2 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fc6c57b763d40d881705ce75d51007f : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef3cb0b4015f48f19c3f46fdd9ba802c : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edfb3c87ea1142fb976a4119d2e82880 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/225d063b224b4efd9cab1d7cd79437ff : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adda03ac069c4e50abf388d6e3aedf58 : SUCCESS in 3m 50s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/acd26e3b239d4603a8e54e78d0e74ce4 : SUCCESS in 53m 15s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/dbed33aa90764c9f83c7d7332e8bf1c0 : SUCCESS in 42m 52s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/6a6b4579359f41fa9e2de5b7dcaad28c : SUCCESS in 52m 11s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/c92b0d122a984ae9840b835c98502d17 : SUCCESS in 43m 52s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/fc85822cc64a4b8ca5588c15fd5b0c11 : SUCCESS in 40m 07s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/a010b987e45e4f0e8a04af49c223c058 : FAILURE in 49m 48s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/5bd3eb51a1f04b109e788f1d1bb24789 : FAILURE in 42m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b00731247874872b09ac317eadb812b : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"763fb880b70dc17bb8466dcfda8af98e468ac852","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-11-04 18:08:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aea9a87cefa1fec87c5f684d7ba002b731c5f7e7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-04 19:21:28.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"8115fc2879157dee7a3f5da37a6e4f0aeaf2da67","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-04 19:24:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dcf59b4b32d9bc5faa55e5df8750dc017f16cc43","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-04 19:43:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"268338abc9e528115d8dad11b5a039215265dba8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 20:30:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/768a7e4fe78e4d588bb7c3887a2c5151 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2abfc2b19c464d20b4c0a7a36450a198 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15aa9900050e437cbf359734b68aa92c : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/860f43a1490a4556a4f3511d2157545a : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/066bc6971bde401cb16c825c10d525cc : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26b7a68f4e074964aa7d18c6fbe8bda7 : SUCCESS in 4m 11s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/f520c2987c1041dab559f435560a2cbb : SUCCESS in 43m 17s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/79bab1b24f8f42c5987e4d35f8513473 : SUCCESS in 42m 58s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/9b6cc7de830f44fa97d26d606c21ce6f : SUCCESS in 41m 28s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/d5387b635aac47888f90a18ef5ef4301 : SUCCESS in 38m 12s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/8e8a324ad2df4d67bc79475489a5da8d : SUCCESS in 36m 51s\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/d05fe12580024b5992f38a774b14ce51 : FAILURE in 44m 16s (non-voting)\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/3a830037219347dabbf89cf4016b2465 : FAILURE in 44m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42e117491b514c919c3c06a9b8ccbbff : SUCCESS in 7m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0ed0b98006c16fca04b845716364866796a5818f","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-11-04 23:48:14.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"f6d6bce7db3c17e2fa7dab3c985102c0a4eba310","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-11-05 02:01:34.000000000","message":"Patch Set 7: Code-Review+1\n\n\u003e Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"59abe4d33a269bc793b7d311efefa603deca5283","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-11-05 02:27:42.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"31bafd789c5fe0c67474c2b59d4da2eec264f72c","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-11-11 05:54:54.000000000","message":"Patch Set 7:\n\nregate","accounts_in_message":[],"_revision_number":7},{"id":"3402e4cf96bcee8972350fd481b6580a848e5fee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 20:20:52.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"35ba930032aff7f8b1a2676fd32925a836263d97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 21:27:13.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4060eb7946b49f08109a263dce2afeb : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9fce475ed8644098cac3a86270a47a8 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/397219938edb4c298e3954b9e3b4cd59 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92ee7a1d61cc4698a47adb27f41a128c : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a6e2af66e814793aa7be84d5475a265 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e7abd95cd384e389ceb8eb68c38d79a : SUCCESS in 5m 00s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/72d328fd5a09470b9c75a8f1ae3d1ab8 : SUCCESS in 1h 01m 12s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/0d691fc53deb434099f99f7aada766b1 : SUCCESS in 50m 53s\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/2499219098964c48a04d55783ff781a1 : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":7},{"id":"dcfbf3f5bad2418b97944b657609e094f09fbb2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 21:27:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"2250819c5ffa35ae80be306d8cbc3a9f79416a94","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 21:28:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d2f34f89904562a441f2fba8a12fba : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"3cb8f2bf3e8283af3b1b7b41ea46dd8602083781","revisions":{"22e32668c0c29f42303a586caafb65783f44e2ff":{"kind":"REWORK","_number":1,"created":"2019-11-03 01:56:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/692691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/91/692691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/91/692691/1"}}},"commit":{"parents":[{"commit":"5b8e0abc998a349ea4fe4c80ef5c518fb54e4878","subject":"Split engine service into three services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/5b8e0abc998a349ea4fe4c80ef5c518fb54e4878"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-03 01:55:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-03 01:55:57.000000000","tz":-420},"subject":"Standardize worker / thread config","message":"Standardize worker / thread config\n\nMove all worker and thread configuration\nto their individual sections in the config.\n\nChange-Id: Ieaec56e66e6804c09827f39bf68d9ec935c42183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/22e32668c0c29f42303a586caafb65783f44e2ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/22e32668c0c29f42303a586caafb65783f44e2ff"}]},"branch":"refs/heads/master"},"2bfc195d91b2cc05a22fbced5a92033083954ba7":{"kind":"REWORK","_number":2,"created":"2019-11-03 02:01:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/692691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/91/692691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/91/692691/2"}}},"commit":{"parents":[{"commit":"5b8e0abc998a349ea4fe4c80ef5c518fb54e4878","subject":"Split engine service into three services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/5b8e0abc998a349ea4fe4c80ef5c518fb54e4878"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-03 01:55:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-03 02:00:53.000000000","tz":-420},"subject":"Standardize worker / thread config","message":"Standardize worker / thread config\n\nMove all worker and thread configuration\nto their individual sections in the config.\n\nChange-Id: Ieaec56e66e6804c09827f39bf68d9ec935c42183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/2bfc195d91b2cc05a22fbced5a92033083954ba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/2bfc195d91b2cc05a22fbced5a92033083954ba7"}]},"branch":"refs/heads/master"},"ae2aa04f148568075c8c070361a276e15ffb5438":{"kind":"REWORK","_number":3,"created":"2019-11-03 02:53:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/692691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/91/692691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/91/692691/3"}}},"commit":{"parents":[{"commit":"5b8e0abc998a349ea4fe4c80ef5c518fb54e4878","subject":"Split engine service into three services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/5b8e0abc998a349ea4fe4c80ef5c518fb54e4878"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-03 01:55:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-03 02:53:26.000000000","tz":-420},"subject":"Standardize worker / thread config","message":"Standardize worker / thread config\n\nMove all worker and thread configuration\nto their individual sections in the config.\n\nChange-Id: Ieaec56e66e6804c09827f39bf68d9ec935c42183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ae2aa04f148568075c8c070361a276e15ffb5438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/ae2aa04f148568075c8c070361a276e15ffb5438"}]},"branch":"refs/heads/master"},"895dda0f1727b74b1157dcbe8f6144a95f1c7bb2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-11-03 09:00:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/692691/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/91/692691/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/91/692691/4"}}},"commit":{"parents":[{"commit":"5b8e0abc998a349ea4fe4c80ef5c518fb54e4878","subject":"Split engine service into three services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/5b8e0abc998a349ea4fe4c80ef5c518fb54e4878"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-03 01:55:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-03 09:00:18.000000000","tz":0},"subject":"Standardize worker / thread config","message":"Standardize worker / thread config\n\nMoved all worker and thread configuration\nto their individual sections in the config,\nand deprecated the old options.\n\nChange-Id: Ieaec56e66e6804c09827f39bf68d9ec935c42183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/895dda0f1727b74b1157dcbe8f6144a95f1c7bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/895dda0f1727b74b1157dcbe8f6144a95f1c7bb2"}]},"branch":"refs/heads/master"},"515089e1c0eac575e5d353bf097ba0acdfcb6f84":{"kind":"REWORK","_number":5,"created":"2019-11-04 19:21:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/692691/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/91/692691/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/91/692691/5"}}},"commit":{"parents":[{"commit":"b93ed46343b34687810c8e0cd19e3c239fc9a9a5","subject":"Split engine service into three services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/b93ed46343b34687810c8e0cd19e3c239fc9a9a5"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-03 01:55:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-04 19:21:40.000000000","tz":-480},"subject":"Standardize worker / thread config","message":"Standardize worker / thread config\n\nMoved all worker and thread configuration\nto their individual sections in the config,\nand deprecated the old options.\n\nChange-Id: Ieaec56e66e6804c09827f39bf68d9ec935c42183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/515089e1c0eac575e5d353bf097ba0acdfcb6f84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/515089e1c0eac575e5d353bf097ba0acdfcb6f84"}]},"branch":"refs/heads/master"},"d0bfab7c6aadb8a9729aa3751e3af5fe5d59ea8e":{"kind":"REWORK","_number":6,"created":"2019-11-04 19:24:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/692691/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/91/692691/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/91/692691/6"}}},"commit":{"parents":[{"commit":"b93ed46343b34687810c8e0cd19e3c239fc9a9a5","subject":"Split engine service into three services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/b93ed46343b34687810c8e0cd19e3c239fc9a9a5"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-03 01:55:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-04 19:24:27.000000000","tz":-480},"subject":"Standardize worker / thread config","message":"Standardize worker / thread config\n\nMoved all worker and thread configuration\nto their individual sections in the config,\nand deprecated the old options.\n\nChange-Id: Ieaec56e66e6804c09827f39bf68d9ec935c42183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d0bfab7c6aadb8a9729aa3751e3af5fe5d59ea8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d0bfab7c6aadb8a9729aa3751e3af5fe5d59ea8e"}]},"branch":"refs/heads/master"},"3cb8f2bf3e8283af3b1b7b41ea46dd8602083781":{"kind":"REWORK","_number":7,"created":"2019-11-04 19:43:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/692691/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/91/692691/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/91/692691/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/91/692691/7"}}},"commit":{"parents":[{"commit":"54997a14ea2f3f2817fb8066d3e691e1d2880ed1","subject":"Split engine service into three services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/54997a14ea2f3f2817fb8066d3e691e1d2880ed1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-03 01:55:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-04 19:43:46.000000000","tz":-480},"subject":"Standardize worker / thread config","message":"Standardize worker / thread config\n\nMoved all worker and thread configuration\nto their individual sections in the config,\nand deprecated the old options.\n\nChange-Id: Ieaec56e66e6804c09827f39bf68d9ec935c42183\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/3cb8f2bf3e8283af3b1b7b41ea46dd8602083781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/3cb8f2bf3e8283af3b1b7b41ea46dd8602083781"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
