)]}'
{"id":"openstack%2Fsenlin~694701","triplet_id":"openstack%2Fsenlin~master~I0377788d0db0e60d44c35f1d26eefd58502a452a","project":"openstack/senlin","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I0377788d0db0e60d44c35f1d26eefd58502a452a","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2019-11-17 17:48:55.000000000","updated":"2019-11-21 17:46:11.000000000","submitted":"2019-11-21 17:42:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694701-1574358167757-7789a045","meta_rev_id":"f8e083c2d490b2c0eb3d5c072cf7d13b8d97de33","_number":694701,"virtual_id_number":694701,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"date":"2019-11-21 15:49:58.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 17:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"all":[{"value":2,"date":"2019-11-19 17:21:13.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-19 01:25:51.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"all":[{"value":1,"date":"2019-11-19 17:21:13.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 01:25:51.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-11-19 17:21:13.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2019-11-21 17:42:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d05c222a3eef9c1f9d0d24dcdb68d5d7cfad3b20","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-17 17:48:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc5390ecbbdbf0afd77aad30ef1dd3c5c41d57ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 18:37:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/571dff119d444e22b62508618960cc04 : SUCCESS in 3m 47s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e05ba6944b9d496eb51d83f541185d48 : SUCCESS in 2m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4701a11cd5554bc8a83d1e58ffa8249b : SUCCESS in 2m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d178975bbd894434975643c0ab607472 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56b31a251de47c99a15100eb93d012f : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9fca02eafc64353a2bd779d6e57b7a0 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb60b798dd854555a540432b6df6eab5 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/407b0f90b0fb4030a341e05221a76d08 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdef492d9e0a4a34b1d3ede580197d25 : SUCCESS in 9m 58s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/81b39b410e6f434e8fb6edd6dea40b82 : SUCCESS in 41m 11s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/a343558f09d1492194a599daf26b8838 : SUCCESS in 44m 25s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/882cb4958c2347568f9aa1824d880676 : SUCCESS in 36m 08s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/570eb8191dfd4e9484ab20dc2c1fe79b : FAILURE in 39m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b483a62de9314f558f1b52ef64960bc9 : SUCCESS in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6ed2c085cc631dffe3c6e91c935a563d9aacc77","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-19 01:25:51.000000000","message":"Patch Set 1: Code-Review+2\n\nOur CI is currently broken without this.","accounts_in_message":[],"_revision_number":1},{"id":"58db6017f9a16ac79c58809164eeb85fdfdb74fc","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-11-19 17:21:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f841ead83a054f89f7b2d15bf0b7493c1e56817","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-19 20:02:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62b05f15434ee70a126f2cdf4ed167f19ef62aa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 21:02:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1fd29ea397294278ba546737d445751f : SUCCESS in 3m 44s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/09d6a1327b184f60a50d972987993e9c : SUCCESS in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d746679314bc43e19511a7ee67f60ea2 : SUCCESS in 2m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6dd2becfb5b43e3b3a1b19d8941ba7a : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e7bd807e374166aed4a81be8095e7c : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5f602171f5448c0a20fd805aab8ff29 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ab57765463a489097b85f47c972a09f : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a60f557ca14d5c88b556a7f21cc699 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ddfd8b4e7d245d091e2e8b16a2b8d31 : SUCCESS in 9m 34s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/29ee0406697d423b8cade2211eeca7b3 : SUCCESS in 51m 20s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/9ebb1ac892fb430aa68aa4c96d4fb0e1 : SUCCESS in 50m 40s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/93c14e91f8b84ad283af0d9d597aa0ed : SUCCESS in 40m 29s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/e0b76844d4134fbc95985afd6ec3d6fb : FAILURE in 45m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a59f4ea3cdbf4114a9ccf43fe47c6240 : SUCCESS in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ff6ff62a1124c7bb43660492a03a26d99f02a27","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-21 15:49:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"625e43fa4d8c1738e8263e29c9deeb986b0ae912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:48:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1895f8d03e3344e9a533a4cca313045b : SUCCESS in 3m 53s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7d2df1528f9c4a9ca75c42dd64b6b35a : SUCCESS in 2m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/35634c9928a440359b07f1a45702ea10 : SUCCESS in 3m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebb83032f9d84e859c780f85af9e9f08 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7c003662194132b3c4eb34a777fc47 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46b539a2698445698d55975e7efd891a : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51ae0fd60bbd4fa9b3d5940533c06a2e : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c138182a1e4084894ca18496b52807 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da153873e262448b88fefcda366650a3 : SUCCESS in 9m 52s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/b7d0ab1e1c9d411f85413cda21ffa9b3 : SUCCESS in 48m 04s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/18c24960f7b142c296b42dc88462e6b5 : SUCCESS in 49m 44s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/16e678c5d7b844c28b6d499b8349c3b6 : SUCCESS in 49m 21s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/aeab5863b8fa4ae286d872a6f61e05b1 : FAILURE in 50m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/663e00b846b64133959a2964cbebfc30 : SUCCESS in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0cccfa226cdfa509e0a6b65d3a99645de0e81b1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:48:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16a02bf7d79929050719fc784c8727edf797e8f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 17:42:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1cb19fe33f1a4ac78d5efaf2d10c4a19 : SUCCESS in 4m 16s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1d3e455cfd4d46a28df6216a544c42b7 : SUCCESS in 2m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/746a75aaf5c04d568629181ed469d0aa : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/153778891e6344ef9d42598b5b1b385f : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef0cc7c9191f4204b7190cefba870803 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2f02c4e4c1c47089dedeb9339917274 : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ffdf816a1aa4c4ea3a7a64d7dfde965 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c686fbd4352044bd96d4f891a03240be : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48bc5b94812540b6aea830e88de6af26 : SUCCESS in 9m 51s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/b45f57806e604f728c622541a84e8dd3 : SUCCESS in 47m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ddc76cdadef4cbf5c039ef878a8938c52ad79a5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 17:42:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2abb4f036944ea153bbd6716515198b867991f6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 17:46:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/baa184489311449b9462b25780172262 : SUCCESS in 1m 23s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1afcdba8d9bd4c5695e63585a1bd919a : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc53a9acc2464c32b470bb6ea24be8c5 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29c0e6fbdfb8e91d0c2bfc6e4fe3f0c317b5504f","revisions":{"29c0e6fbdfb8e91d0c2bfc6e4fe3f0c317b5504f":{"kind":"REWORK","_number":1,"created":"2019-11-17 17:48:55.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/01/694701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/01/694701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/694701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/694701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/01/694701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/01/694701/1"}}},"commit":{"parents":[{"commit":"9523029e026fb71f3f09bcc6542a178c7b6f19c0","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/9523029e026fb71f3f09bcc6542a178c7b6f19c0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-17 17:48:21.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-17 17:48:53.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nSenlin is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\nChange-Id: I0377788d0db0e60d44c35f1d26eefd58502a452a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/29c0e6fbdfb8e91d0c2bfc6e4fe3f0c317b5504f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/29c0e6fbdfb8e91d0c2bfc6e4fe3f0c317b5504f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
