)]}'
{"id":"openstack%2Fsenlin~695628","triplet_id":"openstack%2Fsenlin~master~I9c7b5757a0c0e969f32f60711474486d9e6a28fa","project":"openstack/senlin","branch":"master","topic":"test3","hashtags":[],"change_id":"I9c7b5757a0c0e969f32f60711474486d9e6a28fa","subject":"Always close openstacksdk connections when done","status":"ABANDONED","created":"2019-11-22 05:54:04.000000000","updated":"2019-11-22 17:21:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"174e5c86f838b615c060e13633fce4db1dd73e49","_number":695628,"virtual_id_number":695628,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-22 06:45:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-11-22 06:54:00.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 06:45:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c73d3fd1f3a96e9edbde84cb875e3736b8ae1711","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-22 05:54:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de98b2876a9bbbcf7fd02fcaff724ec2e499120d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 06:45:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b68c8dcec0974710831d2b0519034393 : FAILURE in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e768b3e7ac324c989dafced403ae6c25 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a86f07f0fafd4288a06e1c4befe0a7a2 : FAILURE in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10159692307f465fa8627ed97062281d : FAILURE in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32344912289b4484b414c5d287a813a5 : SUCCESS in 4m 24s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/2596ce5f2a6742d9ada0d2bdca93e9fe : SUCCESS in 43m 40s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/7eba3160d94e4786a32b9967e4e261a0 : SUCCESS in 44m 28s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/c3eb6ff6eb9340e4826c642e0e806126 : SUCCESS in 36m 26s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/2145c4e8f7e24706b4f80ee3e950c3f0 : FAILURE in 43m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b77e00e20f65462d8b82a95c287151ec : FAILURE in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8583f72e626a332569fd22fa07ded7ed28659e38","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-22 06:54:00.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"3335cc204f56529ffcb9d899f30a6f671bb422b4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-22 17:21:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"114ae69332ea345c8e9fe998b615b1d2f0e2a924","revisions":{"114ae69332ea345c8e9fe998b615b1d2f0e2a924":{"kind":"REWORK","_number":1,"created":"2019-11-22 05:54:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/28/695628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/28/695628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/28/695628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/28/695628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/28/695628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/28/695628/1"}}},"commit":{"parents":[{"commit":"29c0e6fbdfb8e91d0c2bfc6e4fe3f0c317b5504f","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/29c0e6fbdfb8e91d0c2bfc6e4fe3f0c317b5504f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-22 05:48:01.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-22 05:53:50.000000000","tz":-480},"subject":"Always close openstacksdk connections when done","message":"Always close openstacksdk connections when done\n\nChange-Id: I9c7b5757a0c0e969f32f60711474486d9e6a28fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/114ae69332ea345c8e9fe998b615b1d2f0e2a924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/114ae69332ea345c8e9fe998b615b1d2f0e2a924"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
