)]}'
{"id":"openstack%2Fsenlin~696792","triplet_id":"openstack%2Fsenlin~master~I95a8b5791eee5d334f8841bebe52f37f6f5eb2e5","project":"openstack/senlin","branch":"master","topic":"fix_resource_leak","hashtags":[],"change_id":"I95a8b5791eee5d334f8841bebe52f37f6f5eb2e5","subject":"Fix leaking resources during cluster recover","status":"MERGED","created":"2019-12-01 09:14:49.000000000","updated":"2019-12-12 00:18:07.000000000","submitted":"2019-12-12 00:15:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696792-1576109755785-758459c9","meta_rev_id":"40ccab7d4eb676f3c1331d3b143058d890fc5621","_number":696792,"virtual_id_number":696792,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 00:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"all":[{"value":2,"date":"2019-12-11 19:34:15.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"all":[{"value":1,"date":"2019-12-11 23:24:21.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 23:24:21.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2019-12-12 00:15:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"032bc8ac00e110fa67c636013bb9c07a994689ce","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-01 09:14:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b591debb85f7d2763de688b3258af7c3df058152","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-01 09:36:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed64815d75253e816db82d5ae899700f27b25f71","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-01 10:15:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a6d15217bf24d2aee5c72875cf7932f1a381139e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-01 10:27:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3192dd09e3868ed3367095e38882e2af5a025b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 11:23:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70721feab4ca48a290226acf9525fd83 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c78e2131bdef44d6900eaffff61915c0 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c004752302db44439e2320acccf11ac9 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9876d4dcacf4205b81d4c17326b5823 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a3656893234819afe0001c0c66037d : SUCCESS in 4m 29s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/38a3d7b0c7cb4c329d56406ed9c87054 : SUCCESS in 46m 25s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/9de6b7d0b97d482593b88e12c0fc5995 : SUCCESS in 47m 24s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/13436d1e24db40e2b4b27a49b5e51115 : SUCCESS in 37m 42s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/acd76dee3185427ba0d91b86228a5321 : FAILURE in 44m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7483dec99cc444fbdcef6caef37a852 : FAILURE in 6m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"675343bed0687ff1eaed4de8105e019981f71981","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-01 18:35:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8383fc3b1d372c110c2eebaddbb160ef555fcea5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-01 18:36:05.000000000","message":"Topic changed from fix_nova_show to fix_resource_leak","accounts_in_message":[],"_revision_number":5},{"id":"2ac69bea854bdc7eae972d658140c545ab5fa84a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-01 18:41:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1e510c1a05abcaae6747ecbd6abcf733ee819311","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 19:19:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d77daac772a74e2a8b1349ebc608d1cc : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df9d51e41db441869831a502c44bafb2 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4814e464f0c4742a77d5bbef5d3fad5 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d43a17dab4884841bf1914138e23dffc : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69253ab95aa41dbb492889fc434fd99 : SUCCESS in 4m 05s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/a6c22a61af0341c9a4c5830d2350df63 : SUCCESS in 41m 33s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/1ed0f1b73355461b80ef44e785402e89 : SUCCESS in 42m 02s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/f206360c65514cfcbc1155d21d28b06c : SUCCESS in 37m 25s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/4e5506e1ec9b4224a9c0fffd5f60b4ff : SUCCESS in 42m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04807952efaa433cac94a9a11864b124 : FAILURE in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f946fd58ae540d05d9c1f2089a130c087cf80eac","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-01 22:56:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"956f6afc249682308dc6fc8d88dfb4c084517c28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 23:52:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2267072f9781412098149942a11b1eb3 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3228efd42784c88adc71b46b9184a4c : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd465735d0004acaa1ea0ad4334b3301 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd0460b8138a46b2a20ef5933866bc3e : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e3d8f7ab15a436db4f3d64ecd7363b8 : SUCCESS in 3m 40s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/17787ffe49424026a8827f5473e611b7 : SUCCESS in 47m 42s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/15704690cbed4af6a2614b8b417799e3 : SUCCESS in 50m 11s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/ceef35764e3342a0ab5a25b7982f3f98 : SUCCESS in 37m 43s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/1594302e8ec3491eb479a2b09c071245 : SUCCESS in 43m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e6451da190e4e5bb39a1daa875d7001 : FAILURE in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cddf4cefb3afd02f1fce0e187af9c448b0558fa6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-02 01:33:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4f7dc48881c8d9cd18a3fe860ad57b47c3a8370a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-02 01:56:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0d8beb9eef541c856e397bbcb7acdc502d81292f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 02:46:40.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6328c7dcd64421eaddcfe21ea00974d : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47033600d23a4d8e9b960ecbf8a99881 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e11f02dfae6342e79928650c5b16202c : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34c3c7ab8e3743ccb9a51058340ece6d : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/883066bad4aa4b929a5038ab0a03086e : SUCCESS in 3m 53s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/df410539d9f342d2b55b756b15894f30 : SUCCESS in 44m 18s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/daf9952e20194454a82f2103ec31feb2 : SUCCESS in 41m 53s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/82c6286df8974d2fbe4b7fb0045efbcd : SUCCESS in 43m 46s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/89369d4473f7440bba04fc9e83210df0 : SUCCESS in 40m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/028aa29c5c8f407a8ce389034962dbf5 : SUCCESS in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"aee4818cef7bf324db8caa2653ce03333033d2e4","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-12-11 19:34:15.000000000","message":"Patch Set 8: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":8},{"id":"1014ff3b86baa0549b35889c9f19603ae3f369c6","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2019-12-11 23:24:21.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"d043668bd47c466028f554db817544f89e034231","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 23:24:33.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"726154957b38729c0a672c2b9c9b8610074709f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 00:15:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"eccb47cdd511b242a32dafeb1254f81ea3fea68c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 00:15:55.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/638162380dd54c3488f53f5f54dc7421 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/958b207476b4421c818ef84e6c0a6a7b : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a56a593c9024fd3a5d790554acabb49 : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/371f28e1caf84ec69cd38897e24c4e3c : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe40a129a2e94128beb1d170d9abf421 : SUCCESS in 5m 15s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/3ee3b9fd88cb4957b50d9000ce98b7b5 : SUCCESS in 48m 46s","accounts_in_message":[],"_revision_number":8},{"id":"40ccab7d4eb676f3c1331d3b143058d890fc5621","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 00:18:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa83eff6cac4c2c92a638115682a5d0 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a6a18b8df5efb179084bf20fe382af452eb79b61","revisions":{"87fd4df7f38326733a0d76672b28d39076fbe573":{"kind":"REWORK","_number":1,"created":"2019-12-01 09:14:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/92/696792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/92/696792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/92/696792/1"}}},"commit":{"parents":[{"commit":"7e04d4a35392ab8c08b6c86b58327bb8b212b15e","subject":"Make sure we always populate flavor_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7e04d4a35392ab8c08b6c86b58327bb8b212b15e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 09:10:41.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 09:14:29.000000000","tz":-480},"subject":"Fix leaking ports during cluster delete","message":"Fix leaking ports during cluster delete\n\nIf a node recover was in progress when a cluster delete was\ncalled it would cause port create to succeed, but unable\nto update the Node object as it no longer exists.\n\nWith this patch if Node.update fails because the object\nno longer exists. Delete all internal ports assosicated\nwith that Node.\n\nChange-Id: I95a8b5791eee5d334f8841bebe52f37f6f5eb2e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/87fd4df7f38326733a0d76672b28d39076fbe573"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/87fd4df7f38326733a0d76672b28d39076fbe573"}]},"branch":"refs/heads/master"},"72c1e94cde33f13d5974c60acb03b54b89861564":{"kind":"REWORK","_number":2,"created":"2019-12-01 09:36:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/92/696792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/92/696792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/92/696792/2"}}},"commit":{"parents":[{"commit":"7e04d4a35392ab8c08b6c86b58327bb8b212b15e","subject":"Make sure we always populate flavor_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7e04d4a35392ab8c08b6c86b58327bb8b212b15e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 09:10:41.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 09:35:55.000000000","tz":-480},"subject":"Fix leaking resources during cluster recover","message":"Fix leaking resources during cluster recover\n\nIf a node recover was in progress when a cluster delete was\ncalled it would cause resources to be created, but unable\nto update the Node object as it no longer exists.\n\nWith this patch if Node.update fails because the object\nno longer exists. Delete all resources created during\nthe failed operation.\n\nChange-Id: I95a8b5791eee5d334f8841bebe52f37f6f5eb2e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/72c1e94cde33f13d5974c60acb03b54b89861564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/72c1e94cde33f13d5974c60acb03b54b89861564"}]},"branch":"refs/heads/master"},"0939c7e8c8ed83f93b3932053c47c08fafb0ac16":{"kind":"REWORK","_number":3,"created":"2019-12-01 10:15:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/92/696792/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/92/696792/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/92/696792/3"}}},"commit":{"parents":[{"commit":"7e04d4a35392ab8c08b6c86b58327bb8b212b15e","subject":"Make sure we always populate flavor_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7e04d4a35392ab8c08b6c86b58327bb8b212b15e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 09:10:41.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 10:15:31.000000000","tz":-480},"subject":"Fix leaking resources during cluster recover","message":"Fix leaking resources during cluster recover\n\nIf a node recover was in progress when a cluster delete was\ncalled it would cause resources to be created, but unable\nto update the Node object as it no longer exists.\n\nWith this patch if Node.update fails because the object\nno longer exists. Delete all resources created during\nthe failed operation.\n\nChange-Id: I95a8b5791eee5d334f8841bebe52f37f6f5eb2e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/0939c7e8c8ed83f93b3932053c47c08fafb0ac16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/0939c7e8c8ed83f93b3932053c47c08fafb0ac16"}]},"branch":"refs/heads/master"},"a00d0c985b7af1f2ece387bbfa3fb84319efdee0":{"kind":"REWORK","_number":4,"created":"2019-12-01 10:27:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/92/696792/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/92/696792/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/92/696792/4"}}},"commit":{"parents":[{"commit":"7e04d4a35392ab8c08b6c86b58327bb8b212b15e","subject":"Make sure we always populate flavor_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7e04d4a35392ab8c08b6c86b58327bb8b212b15e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 09:10:41.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 10:26:47.000000000","tz":-480},"subject":"Fix leaking resources during cluster recover","message":"Fix leaking resources during cluster recover\n\nIf a node recover was in progress when a cluster delete was\ncalled it would cause resources to be created, but unable\nto update the Node object as it no longer exists.\n\nWith this patch if Node.update fails because the object\nno longer exists. Delete all resources created during\nthe failed operation.\n\nChange-Id: I95a8b5791eee5d334f8841bebe52f37f6f5eb2e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/a00d0c985b7af1f2ece387bbfa3fb84319efdee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/a00d0c985b7af1f2ece387bbfa3fb84319efdee0"}]},"branch":"refs/heads/master"},"c5235aaa9aa4ec9485fba1c7cdbe34df2ddf2cdd":{"kind":"REWORK","_number":5,"created":"2019-12-01 18:35:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/92/696792/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/92/696792/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/92/696792/5"}}},"commit":{"parents":[{"commit":"7e04d4a35392ab8c08b6c86b58327bb8b212b15e","subject":"Make sure we always populate flavor_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7e04d4a35392ab8c08b6c86b58327bb8b212b15e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 09:10:41.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 18:35:41.000000000","tz":-480},"subject":"Fix leaking resources during cluster recover","message":"Fix leaking resources during cluster recover\n\nIf a node recover was in progress when a cluster delete was\ncalled it would cause resources to be created, but unable\nto update the Node object as it no longer exists.\n\nWith this patch if Node.update fails because the object\nno longer exists. Delete all resources created during\nthe failed operation.\n\nChange-Id: I95a8b5791eee5d334f8841bebe52f37f6f5eb2e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c5235aaa9aa4ec9485fba1c7cdbe34df2ddf2cdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c5235aaa9aa4ec9485fba1c7cdbe34df2ddf2cdd"}]},"branch":"refs/heads/master"},"e69869396b32d7b8526f6ed6b273861d7dc61802":{"kind":"REWORK","_number":6,"created":"2019-12-01 22:56:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/92/696792/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/92/696792/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/92/696792/6"}}},"commit":{"parents":[{"commit":"7e04d4a35392ab8c08b6c86b58327bb8b212b15e","subject":"Make sure we always populate flavor_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7e04d4a35392ab8c08b6c86b58327bb8b212b15e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 09:10:41.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 22:56:05.000000000","tz":-480},"subject":"Fix leaking resources during cluster recover","message":"Fix leaking resources during cluster recover\n\nIf a node recover was in progress when a cluster delete was\ncalled it would cause resources to be created, but unable\nto update the Node object as it no longer exists.\n\nWith this patch if Node.update fails because the object\nno longer exists. Delete all resources created during\nthe failed operation.\n\nChange-Id: I95a8b5791eee5d334f8841bebe52f37f6f5eb2e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e69869396b32d7b8526f6ed6b273861d7dc61802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/e69869396b32d7b8526f6ed6b273861d7dc61802"}]},"branch":"refs/heads/master"},"9c50b79951903dc642a6ebd63e1c6ea5109c26ef":{"kind":"REWORK","_number":7,"created":"2019-12-02 01:33:27.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/92/696792/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/92/696792/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/92/696792/7"}}},"commit":{"parents":[{"commit":"7e04d4a35392ab8c08b6c86b58327bb8b212b15e","subject":"Make sure we always populate flavor_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7e04d4a35392ab8c08b6c86b58327bb8b212b15e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 09:10:41.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-02 01:30:33.000000000","tz":-480},"subject":"Fix leaking resources during cluster recover","message":"Fix leaking resources during cluster recover\n\nIf a node recover was in progress when a cluster delete was\ncalled it would cause resources to be created, but unable\nto update the Node object as it no longer exists.\n\nWith this patch if Node.update fails because the object\nno longer exists. Delete all resources created during\nthe failed operation.\n\nChange-Id: I95a8b5791eee5d334f8841bebe52f37f6f5eb2e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/9c50b79951903dc642a6ebd63e1c6ea5109c26ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/9c50b79951903dc642a6ebd63e1c6ea5109c26ef"}]},"branch":"refs/heads/master"},"a6a18b8df5efb179084bf20fe382af452eb79b61":{"kind":"REWORK","_number":8,"created":"2019-12-02 01:56:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/92/696792/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/92/696792/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/92/696792/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/92/696792/8"}}},"commit":{"parents":[{"commit":"7e04d4a35392ab8c08b6c86b58327bb8b212b15e","subject":"Make sure we always populate flavor_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/7e04d4a35392ab8c08b6c86b58327bb8b212b15e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-01 09:10:41.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-02 01:56:41.000000000","tz":-480},"subject":"Fix leaking resources during cluster recover","message":"Fix leaking resources during cluster recover\n\nIf a node recover was in progress when a cluster delete was\ncalled it would cause resources to be created, but unable\nto update the Node object as it no longer exists.\n\nWith this patch if Node.update fails because the object\nno longer exists. Delete all resources created during\nthe failed operation.\n\nChange-Id: I95a8b5791eee5d334f8841bebe52f37f6f5eb2e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/a6a18b8df5efb179084bf20fe382af452eb79b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/a6a18b8df5efb179084bf20fe382af452eb79b61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
