)]}'
{"id":"openstack%2Fsenlin~700762","triplet_id":"openstack%2Fsenlin~master~Ie0cbdbec29be99d58a032e77f7087c2bcf08b65d","project":"openstack/senlin","branch":"master","topic":"bug/1857959","hashtags":[],"change_id":"Ie0cbdbec29be99d58a032e77f7087c2bcf08b65d","subject":"Fix error when attach load balancer policy","status":"ABANDONED","created":"2019-12-30 07:52:20.000000000","updated":"2021-11-29 04:46:41.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e29b84dc4e9cff29629163c7883ae8bbdd8d2084","_number":700762,"virtual_id_number":700762,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"date":"2020-01-17 02:22:59.000000000","_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"date":"2020-01-16 19:52:52.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-15 05:33:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"date":"2020-01-18 06:06:51.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-07 19:50:40.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-01-08 01:33:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-01-08 05:34:41.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2020-01-15 05:33:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-16 19:52:52.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2020-01-18 06:06:51.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"4dc59cddf0b6d22c49b8d076969bf991a83a457a","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-12-30 07:52:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd25fc5d3f5f360c318568a7b654592b96ae6006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 09:10:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4fa8136ae1a4d52965d22a6ac4e1d23 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e769ca1dc8d431ebad3f60f3406c9f1 : FAILURE in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3580f9e4c21948e580f4892680c748c3 : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/565dcbf8e3e144e3a9215f2353532863 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d708b5a7994d41bdb652ecdf4d430f75 : SUCCESS in 4m 33s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/09ed53efb10248029f5094e8724f78a2 : SUCCESS in 48m 10s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/66248eb433b24b459191406259d4ee8e : SUCCESS in 59m 25s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/8c04c98b57ee4a85b1c5dab454487524 : SUCCESS in 48m 51s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/2d0118db654b43648d05861bf85ef634 : SUCCESS in 43m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e772b6ed9a9a4a16843e0f8ab47549db : FAILURE in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50c65a2e6f8ff721c10c1cba7d266362b430ff25","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-12-30 09:19:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3309aa0affc66bf446f8e055221e50dc011c6dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 10:22:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1777b4ad0200470ab711d9a361f2dc62 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eeade590fec4beba43998e8a9cf32be : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b9ea42cf84c46819d629947552d0a79 : SUCCESS in 6m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10c10ad8b2a74e37970854b9b7a7e0b1 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab008673f6c64a8292419b12ad0a953a : SUCCESS in 3m 49s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/1445e38ec7b64c2a925555f5eb63b0d9 : FAILURE in 59m 08s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/a3f09328a1804e358dcdb8f645773047 : FAILURE in 59m 15s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/f60eade284e34eac91dfd9cd0a500170 : SUCCESS in 42m 57s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/1038a3c7b82640b1b3786081972ef089 : SUCCESS in 47m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76d3f6ad95a2423ab8ac3fe0df0dc7b3 : FAILURE in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"856dc1d070de21f1a1416f2840d01c5233b2bc06","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-12-31 02:56:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b68825eda9fef62ff0ea76b385537864833e6ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 03:50:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9716c554af5c42e389220844c862172a : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a182c33c32ed484282958202322da75c : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f78d49a41294a3993df01cb3f6e213e : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5908e007310f4ff79cf5358fb81596b1 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f13a0b5ada24416bc6a93661072b254 : SUCCESS in 3m 41s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/510e3c5d7a484b0a9c26680c14d5d703 : SUCCESS in 44m 05s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/54958e497b194d6cbbe21a09d5cf6c65 : SUCCESS in 40m 45s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/38cbc11a6c7547df838fbad6180ac107 : SUCCESS in 38m 30s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/6aab568247f6449eaf05c2acace1ab82 : SUCCESS in 47m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a0de17cf1c348e2b6410df316dafdf8 : FAILURE in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"98bf2f77970286f21a75c0feebffb879c3d43828","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-31 04:06:50.000000000","message":"Patch Set 2:\n\nI haven\u0027t had time to take a look yet, but could you add a bug report and link it to this commit? I also looped the Octavia people for additional help reviewing this.","accounts_in_message":[],"_revision_number":2},{"id":"51e0bcb17d1706e3f82fe3eb65ac6b03b0819477","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-12-31 07:59:40.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"787333387d63d1328f23ccf4771d0f6fbf795a30","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2019-12-31 08:03:18.000000000","message":"Patch Set 3:\n\n\u003e I haven\u0027t had time to take a look yet, but could you add a bug\n \u003e report and link it to this commit? I also looped the Octavia people\n \u003e for additional help reviewing this.\n\nYep. Thank you. \nHappy New Year!","accounts_in_message":[],"_revision_number":3},{"id":"7c854a91600922000cbfb29370f001a21038d750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 08:57:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11c10508fa1744c49820ccf7828666fe : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d6fa49f6de240de9ad77e7422be5d6f : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/579d110b36e64115adbf68e9be7211f3 : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0447b3fecc104226ac1bb909058bf8c6 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd74ef9f3c874df2979f58ce37307cca : SUCCESS in 6m 06s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/794cc5e8ac434449a4a82b7a003fc65f : SUCCESS in 53m 27s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/92e2b82690cf4ee9931c472692c9e90d : SUCCESS in 51m 13s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/43d78e49087f4f43b9663ac83a209590 : SUCCESS in 47m 07s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/c861a72156c546478e8e164fd468b3ac : SUCCESS in 53m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d26f9387478746ffbdaefce93d824fdd : FAILURE in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1391b63fce3df3013aefb633d1040f3e19c135f1","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2020-01-07 00:59:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4d45e372eb46498e7bb6e6bc0f80396899e772c3","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2020-01-07 03:52:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bd577f79227522af435260e5c957facce3c40ba8","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2020-01-07 19:41:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3d7b461205e9c7fddc1d8461eeaafe770029be75","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-01-07 19:50:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nPer Duc\u0027s comments, this should be revised","accounts_in_message":[],"_revision_number":3},{"id":"115d31d111e0d7eba8724c3648f4dd470ba215c2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-08 01:33:25.000000000","message":"Patch Set 3:\n\nI\u0027m not sure I fully understand the issue here.\nI think what is happening is you are attempting to add members to a pool that have duplicate IP:PORT pairs to existing members.\nThis is not a valid load balancer configuration as there should only be one member per backend service endpoint.\n\nIf you are simply trying to update the members of a pool(create/update/remove) with a new known-good list, you might consider using the batch member update capabilities. https://docs.openstack.org/api-ref/load-balancer/v2/index.html#batch-update-members\nThough it looks like that feature may not have made it into openstacksdk yet.","accounts_in_message":[],"_revision_number":3},{"id":"d645d41454d9aa90cc74c5fb31d8f18b9b06555e","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2020-01-08 02:05:50.000000000","message":"Patch Set 3:\n\n\u003e I\u0027m not sure I fully understand the issue here.\n \u003e I think what is happening is you are attempting to add members to a\n \u003e pool that have duplicate IP:PORT pairs to existing members.\n \u003e This is not a valid load balancer configuration as there should\n \u003e only be one member per backend service endpoint.\n \u003e \n \u003e If you are simply trying to update the members of a\n \u003e pool(create/update/remove) with a new known-good list, you might\n \u003e consider using the batch member update capabilities.\n \u003e https://docs.openstack.org/api-ref/load-balancer/v2/index.html#batch-update-members\n \u003e Though it looks like that feature may not have made it into\n \u003e openstacksdk yet.\n\nYes, my problem in here is duplicate many IP:Port like bug description. But, this was created by my client in two other services. So, I want resolve it.","accounts_in_message":[],"_revision_number":3},{"id":"9a9a2e3f2fae89aa56b7896c66cb40686754b72d","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2020-01-08 02:06:15.000000000","message":"Patch Set 3:\n\n\u003e (1 comment)\n\nYep. Thank you.","accounts_in_message":[],"_revision_number":3},{"id":"3b462fe386da576ca0eb909fe80e28729d17ccf3","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2020-01-08 05:34:41.000000000","message":"Patch Set 3: Code-Review-1\n\nopenstack-tox-cover fail: \nERROR:   cover: commands failed\n2019-12-31 08:09:05.368847 | ubuntu-bionic | /usr/local/lib/python3.6/dist-packages/tox/config/__init__.py:582: UserWarning: conflicting basepython version (set 3.6, should be 3.7) for env \u0027py37\u0027;resolve conflict or set ignore_basepython_conflic","accounts_in_message":[],"_revision_number":3},{"id":"7246e5d4c88c3edbaa698574815426c188130aed","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2020-01-15 04:41:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"140a8945999076936d3ed6701a9ba3150ced92f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 05:33:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/52f09be7a8dc4700b1929844e5ee4617 : SUCCESS in 3m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3551f6a738f4311835c2040efbce3d1 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a1649738258455796ec18899891c714 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e79373a37c04127ba214682ac6bc6ef : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0fc534ceeddf49d18167c73b50abc40e : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc20b3171c934fd583c70b54c8bf887f : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2ec1d1ae5e042e5a6eb50ae2549cfb7 : SUCCESS in 10m 14s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/35d2aa4d0ef84c08b54269ddbd86ac64 : SUCCESS in 43m 53s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/aacab424e7c0473689f06681a35c98fc : SUCCESS in 46m 01s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/ee50e0da695d4b08a251d8c711af7721 : SUCCESS in 38m 08s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/d36f28409a884bdeb531f4e66c2c765b : SUCCESS in 44m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58e5a4ff48a546b7b3f5b7acde0594ec : FAILURE in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fe4db0c153113c7cd03b73b6e1f546b8ba655b93","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2020-01-16 19:52:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e0392f792f39038460e17610254ed3936ba1761f","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2020-01-17 02:22:59.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\nNo, I don\u0027t want like its.\nI hope members was existed in pool of LB that are nodes of cluster being ignore when I do attach LB policy to cluster.\nBecause in my work, my user usually add multiple servers(nodes of cluster) even an IP address with multiple ports to pool LB and attach LB policy to cluster. So, the conflict will come.\nAnd, I don\u0027t want like that.\n\nTo resolve this problem, I need check couple IP:Port was existed in pool LB or not? If was existed, I need return ID of member in pool LB.","accounts_in_message":[],"_revision_number":4},{"id":"ec017ff3e06e1c9328941da2b8cb8f746340b7c6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-01-18 06:06:13.000000000","message":"Patch Set 4:\n\n\u003e \u003e (1 comment)\n \u003e \n \u003e No, I don\u0027t want like its.\n \u003e I hope members was existed in pool of LB that are nodes of cluster\n \u003e being ignore when I do attach LB policy to cluster.\n \u003e Because in my work, my user usually add multiple servers(nodes of\n \u003e cluster) even an IP address with multiple ports to pool LB and\n \u003e attach LB policy to cluster. So, the conflict will come.\n \u003e And, I don\u0027t want like that.\n \u003e \n \u003e To resolve this problem, I need check couple IP:Port was existed in\n \u003e pool LB or not? If was existed, I need return ID of member in pool\n \u003e LB.\n\nI think Duc\u0027s version looks a lot cleaner and I can\u0027t see any reason why his code wouldn\u0027t work. \nBo could you add unit test cases to catch this? That way you could use this to prove that that your code catches this and not Duc\u0027s version, plus it better helps us understand the issue in general.","accounts_in_message":[],"_revision_number":4},{"id":"c94895cc98daa87cc7799fcce0364136ba025288","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-01-18 06:06:51.000000000","message":"Patch Set 4:\n\n\u003e \u003e \u003e (1 comment)\n \u003e \u003e\n \u003e \u003e No, I don\u0027t want like its.\n \u003e \u003e I hope members was existed in pool of LB that are nodes of\n \u003e cluster\n \u003e \u003e being ignore when I do attach LB policy to cluster.\n \u003e \u003e Because in my work, my user usually add multiple servers(nodes of\n \u003e \u003e cluster) even an IP address with multiple ports to pool LB and\n \u003e \u003e attach LB policy to cluster. So, the conflict will come.\n \u003e \u003e And, I don\u0027t want like that.\n \u003e \u003e\n \u003e \u003e To resolve this problem, I need check couple IP:Port was existed\n \u003e in\n \u003e \u003e pool LB or not? If was existed, I need return ID of member in\n \u003e pool\n \u003e \u003e LB.\n \u003e \n \u003e I think Duc\u0027s version looks a lot cleaner and I can\u0027t see any\n \u003e reason why his code wouldn\u0027t work.\n \u003e Bo could you add unit test cases to catch this? That way you could\n \u003e use this to prove that that your code catches this and not Duc\u0027s\n \u003e version, plus it better helps us understand the issue in general.\n\ntl;dr add a unit test to help us understand the edge case this is fixing","accounts_in_message":[],"_revision_number":4},{"id":"e29b84dc4e9cff29629163c7883ae8bbdd8d2084","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-29 04:46:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"19bd0299b280ec073090da3dbfe9ee9138f23825","revisions":{"37a5242deac99564832859adda36005a3cf66f0f":{"kind":"REWORK","_number":1,"created":"2019-12-30 07:52:20.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/62/700762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/62/700762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/62/700762/1"}}},"commit":{"parents":[{"commit":"abdc171303cf5a51eab1019099d12a5416ed5a4f","subject":"Update release notes to better reflect new configs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/abdc171303cf5a51eab1019099d12a5416ed5a4f"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-12-30 07:46:50.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-12-30 07:46:50.000000000","tz":420},"subject":"Fix error when attach load balancer policy","message":"Fix error when attach load balancer policy\n\nWhen LB was existed member(s) of a cluster, and we attach policy\nusing this LB and cluster. We will receive error log.\n\nThis patch was resolved this problem\n\nChange-Id: Ie0cbdbec29be99d58a032e77f7087c2bcf08b65d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/37a5242deac99564832859adda36005a3cf66f0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/37a5242deac99564832859adda36005a3cf66f0f"}]},"branch":"refs/heads/master"},"0b66c4734ddbafefe769fc7b19ddf555efa103e9":{"kind":"REWORK","_number":2,"created":"2019-12-30 09:19:52.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/62/700762/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/62/700762/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/62/700762/2"}}},"commit":{"parents":[{"commit":"abdc171303cf5a51eab1019099d12a5416ed5a4f","subject":"Update release notes to better reflect new configs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/abdc171303cf5a51eab1019099d12a5416ed5a4f"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-12-30 07:46:50.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-12-30 09:19:28.000000000","tz":420},"subject":"Fix error when attach load balancer policy","message":"Fix error when attach load balancer policy\n\nWhen LB was existed member(s) of a cluster, and we attach policy\nusing this LB and cluster. We will receive error log.\n\nThis patch was resolved this problem\n\nChange-Id: Ie0cbdbec29be99d58a032e77f7087c2bcf08b65d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/0b66c4734ddbafefe769fc7b19ddf555efa103e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/0b66c4734ddbafefe769fc7b19ddf555efa103e9"}]},"branch":"refs/heads/master"},"1803cc4c27a5a77e39b4567a55b294910729d148":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-12-31 07:59:40.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/62/700762/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/62/700762/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/62/700762/3"}}},"commit":{"parents":[{"commit":"abdc171303cf5a51eab1019099d12a5416ed5a4f","subject":"Update release notes to better reflect new configs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/abdc171303cf5a51eab1019099d12a5416ed5a4f"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-12-30 07:46:50.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-12-31 07:58:49.000000000","tz":420},"subject":"Fix error when attach load balancer policy","message":"Fix error when attach load balancer policy\n\nWhen LB was existed member(s) of a cluster, and we attach policy\nusing this LB and cluster. We will receive error log.\n\nThis patch was resolved this problem\n\nCloses-Bug: 1857959\nChange-Id: Ie0cbdbec29be99d58a032e77f7087c2bcf08b65d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/1803cc4c27a5a77e39b4567a55b294910729d148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/1803cc4c27a5a77e39b4567a55b294910729d148"}]},"branch":"refs/heads/master"},"19bd0299b280ec073090da3dbfe9ee9138f23825":{"kind":"REWORK","_number":4,"created":"2020-01-15 04:41:56.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/62/700762/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/62/700762/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/62/700762/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/62/700762/4"}}},"commit":{"parents":[{"commit":"c83e79644adb3945787b1d3c7ddc1e078e883e4e","subject":"Merge \"Move test driver to the tests folder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/c83e79644adb3945787b1d3c7ddc1e078e883e4e"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-12-30 07:46:50.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2020-01-15 04:41:23.000000000","tz":420},"subject":"Fix error when attach load balancer policy","message":"Fix error when attach load balancer policy\n\nWhen LB was existed member(s) of a cluster, and we attach policy\nusing this LB and cluster. We will receive error log.\n\nThis patch was resolved this problem\n\nCloses-Bug: 1857959\nChange-Id: Ie0cbdbec29be99d58a032e77f7087c2bcf08b65d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/19bd0299b280ec073090da3dbfe9ee9138f23825"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/19bd0299b280ec073090da3dbfe9ee9138f23825"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
