)]}'
{"id":"openstack%2Fsenlin~749874","triplet_id":"openstack%2Fsenlin~master~Ifa475ff146af8b49a762218ff244f0ff0fee9ac0","project":"openstack/senlin","branch":"master","topic":"749874","hashtags":[],"change_id":"Ifa475ff146af8b49a762218ff244f0ff0fee9ac0","subject":"Add options to choose endpoints and SSL verification","status":"MERGED","created":"2020-09-04 06:48:41.000000000","updated":"2021-01-14 15:49:54.000000000","submitted":"2020-10-13 23:30:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749874-1602631850714-eaa21e4c","meta_rev_id":"2dc85d126e2b852268051acf53b05ac51d8c0fb7","_number":749874,"virtual_id_number":749874,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-13 23:30:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"all":[{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":2,"date":"2020-10-13 16:09:35.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":1,"date":"2020-10-13 07:23:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-13 21:16:23.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"all":[{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":1,"date":"2020-10-13 22:34:34.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 15:01:22.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2020-10-13 21:16:23.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-10-13 22:34:34.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2020-10-13 23:30:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-14 15:49:54.000000000","updated_by":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"reviewer":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"state":"CC"}],"messages":[{"id":"cb69013f675f2af10bf73fef169e000a0768073c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-04 06:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5f1989a2a786922cf23f2cb642ee8d2b99b2603","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-04 07:03:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"337c32211e00c8246ff140538e7bf5eff7922d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 08:02:08.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84c8b3d7dd2a426087578e32b1970beb : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27dd132185444eed8a69100389655009 : FAILURE in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8edfe277d3e4437abf9c035d31b06627 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/264013ad980d46e58dc52fc213314ac7 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8803a37702455a953ba5a91e34ae20 : SUCCESS in 5m 47s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/442e71937c374ef68e98bdfc369ceceb : SUCCESS in 50m 02s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/26c5c3473c6847bab7825fd1d3b13d97 : SUCCESS in 49m 03s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/e747ee5316274f319a581356784e1c75 : SUCCESS in 43m 36s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/4bb0f7a631834d7fa1075a42a488b5ff : SUCCESS in 46m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da5484bfcf354234a9e5be345bd22ded : SUCCESS in 8m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa3129ab5db1689c883fbc58bb70c9a65c3c1d3e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-04 08:38:18.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"9b743718cd41c3e140dd0c96ed7b1bed3ed3a1ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 09:36:21.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da5465b72a984b70bf441546d2b4bb12 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b146b34be4c461f900279f2c178ecb9 : FAILURE in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a042ab0b472a4c529af1f18513ae48cb : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0d73795c6f247eca294c3f45c84e33e : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8035b39e1b7949ffbc3fe5c7d2a64474 : SUCCESS in 5m 35s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/acf1d3ef36634a1384ce5e0a94be6751 : SUCCESS in 44m 42s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8747dc6537b64a90b3ad8bd5d6cf7f6c : SUCCESS in 50m 51s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/be06048fea6b429f87f188231cae2fab : SUCCESS in 40m 49s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/dfd2da986f914b91ad6d15e859a24f9c : SUCCESS in 44m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9314178f3e3540ba8c4fb3d4ca5116ed : SUCCESS in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"667e71a3aa4ad9ba48f4e0de6b0c58ef70f1f6cb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-04 09:42:11.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"67fa5e7c4e4875e4f0c83cb0a1318b64df032f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:57:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b24f85724904545bffe7d58e435640b : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/225b58be9a5b4ade924e5ee0b52edef3 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67f89d8e2e2b45f6a6b84aca9d25d391 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1786ae86f6ff4c27a2966ca7abec3509 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f940299b1a334fe49827d7775109f29b : SUCCESS in 4m 22s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/4552ad5e8f014da78c1048188b13ea84 : SUCCESS in 40m 32s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/d357c2cef33043ef9f4b7cf8b6a10ab5 : SUCCESS in 44m 03s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/2e930ff974344b709ebf84ce39038a01 : SUCCESS in 42m 41s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/487c6804c7d74373bd3810338dabd6c0 : SUCCESS in 1h 10m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4267e05ef9a643e7a0f210c7dccee209 : SUCCESS in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c716a1b9f778962d36e746daac6481c0e42ad83d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-14 12:56:25.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"8fab03abe7bb47504ced082061fdd360f3d2bbdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 13:45:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/489939a43ab942d09614a942b855ac46 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d45f9d225f04235bc21d7e50b4fe9b5 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/255867bf503d41e4acd2a7d2edcefad2 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/696ff39194dc4baba8ee2abd07947b00 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccde651880b646cfa842e7b6ffa19ee5 : SUCCESS in 4m 07s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/7d32c88272774037846d41567a63e76b : FAILURE in 46m 38s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/c2856f9be3924f109aa44ca0cee8563c : SUCCESS in 39m 30s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/b5dc3909a95a4c01a5043cd3fa0f34ff : SUCCESS in 30m 22s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/e08dd122cdad456d8052bd0e615c9356 : SUCCESS in 44m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0669061ef2774892bad085e80a2b1153 : SUCCESS in 7m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"57a4d1cb76d028a1a050a37518b2cb0237a00db3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-15 08:46:27.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a58179fa791c28d04f7395945d9525f8b974e9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 09:19:17.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45f966ec669047b7b7e6fca622db0956 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11d2fdd2eed945ccb025763ac0041d76 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/343d05dc66104d1c8f87e6ceb58ad3c6 : FAILURE in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a574a958cf8450dbab5a526850a6c71 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7964637372524bdd8fa05e32ff269277 : SUCCESS in 4m 27s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/b604b421ac564f249e6aff9be6008ef5 : FAILURE in 18m 46s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8567be61d4c64d6f97582addb3c31189 : FAILURE in 16m 01s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/8c45cd2e6f244226945e2afee52655fd : FAILURE in 17m 07s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/73d4a95393194029b00090dbd19934a2 : FAILURE in 15m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bab9e8d3fa4740b2be07ccd27471c14d : SUCCESS in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c8b7c20827d333a7a61d9e68a1c32e6b04ced9b4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-15 11:38:19.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"726e39e80e9b28b2df0863fec67ae93fedae454d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 12:11:54.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f863a2034c847f19b583725800c0bec : FAILURE in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32bd2d81905745aab91eb1d90536de09 : FAILURE in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbd4d8bc9b4943119190057785a90372 : FAILURE in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c14d40a44fc473ab5ee29dcaeb569a7 : FAILURE in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e536cfad24a469c8f2cd0a96448060f : FAILURE in 3m 12s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/82e053badf394d7caeace64392c61fc2 : FAILURE in 24m 17s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/c5fdbbe8523a4b15b6ba3638862ceafa : FAILURE in 15m 03s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/605bdd6c71144ad2abf0390d683cfe05 : FAILURE in 14m 25s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/69d7c3044ef64eb39b80b78a0ae99379 : FAILURE in 14m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b0ea492401d49caa3bdfbdb371c956e : FAILURE in 3m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b6c3eda7f0009eebc6f3390904d7b88d3d7e9739","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-16 13:46:21.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f8b5e5f1f54d51990431c57ebe9f085790d0fe87","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-16 13:46:25.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"87a8dd4809499d4653f33de96a006f06359e6ef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 14:44:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/339863de39484a3899af7bb3a000b32c : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2bec6d6fb2e4f23b1f1ff841d821b1c : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d4c48d01727460fb17354177ad1fa5a : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c1a78737a0d4de499c33ef61070cada : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3041e1c1f340e0aef0e982e8166fd6 : SUCCESS in 5m 09s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/1eae35e3fd5a4537a36a8164e0e3a7ff : SUCCESS in 49m 41s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/28844fb09a544d8f817de1fe46b3cc6d : SUCCESS in 45m 34s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/e57546fa3169433f963d9d4361a86c3c : SUCCESS in 43m 13s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/df7eb44508b744b8ba552430f07bdf3b : SUCCESS in 46m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68f2ac7578224f68a67177c12e40b459 : SUCCESS in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"346989c0924a3d496260fee90319b238e538cf5e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-17 05:41:28.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"58d5e5d185eb7e59b3f67f14846d29e1692c0c87","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2020-10-01 00:44:59.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"97b135f9bc8ae9f43f88a4ef7f3f0fd9bec493b2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-01 07:24:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"50d0c500976d7803439f575118ded31121155474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 08:24:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/737ca3e77e724485996d96bc045caef8 : FAILURE in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f46296f254a4b4fb99be0d1692c4f15 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97c7f736b6a24f4e9571c69c9a65e2ae : FAILURE in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86c4826ba1654bed99b0853f320a408a : FAILURE in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a335128d90a44ef98ac9eb2908a407eb : SUCCESS in 4m 01s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/a8dff7f591a744409bb60f43deef1a15 : FAILURE in 47m 19s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/25609e03cba6411f9e3bcfa14352889c : FAILURE in 47m 01s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/a11bd7a79b8441eaa0ab4803c9dc386d : FAILURE in 41m 27s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/3c98df5f458c48ac83e3b1a86ad83ce4 : SUCCESS in 52m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e17bdd8491a3431abc8ce77eed3834f8 : FAILURE in 6m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"25d18bbf8ee7ed8bc2b9d10faef7c56cbe2c336a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-01 17:28:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cdb35e8cd9cfd3bbfce88dd5fb05b583bccd96a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 18:49:28.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c838afe09ba4e4a8171dfb7925623c2 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0320ce11f9b446e3871d150218de25e5 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcc41647c08f4865af8a8337bd01938c : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c0a195f253d4d1bb9479d8c93ba1e40 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/259f325670e9454b98ce1cc8f4b67e7e : SUCCESS in 7m 44s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/7c02e5e56ad4407b9bb9a56e79431087 : FAILURE in 50m 45s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/2d1eb98b72954c4695182d063b26ae6f : FAILURE in 46m 49s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/f29395181f4c49aeaaaf3055bd0cf2d3 : FAILURE in 44m 22s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/1a4327b6b1794f5a894f2404b9af604c : SUCCESS in 47m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e22274269bf4734a475cfb3a3e85ade : SUCCESS in 7m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"982dca20b4f13b94772245de01c5de7ca1404b11","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 08:53:17.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9375665794a377b0cfe8f24793d19dac1425df38","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 08:54:51.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"69362fa795239bb627f98d8014f2b5b4a933c7f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 09:52:25.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/841af1a1db014712ba4560b076fe4b50 : FAILURE in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b17abeaeff420b9d489201a906bfbe : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03d12195b7ab42bab45c27f3b80f215d : FAILURE in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a2c9af4c2e34bb59e4c2cdfaf036dd1 : FAILURE in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc9d733fb55c427f9f135e489f36ef6c : SUCCESS in 4m 43s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/e059b57b83a74a18a1a24b35ca5919c1 : FAILURE in 52m 42s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/6367007ccdbe4b4fb7f34e0507f8da0c : FAILURE in 49m 21s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/2921d3e9e1554e35822523daa496dbef : FAILURE in 47m 08s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/4a9bd9d3dea041a0ab06a911fe97f4ef : SUCCESS in 43m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ae8595b17784348a361fcb83154cd05 : FAILURE in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"20b1c025ab69d3b9388781602ac9a85d905a1e0c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 18:16:15.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3187acdaf20fcae663c7ba7d1a39363e6d860751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 19:17:57.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55732c14a0c841ee83a6c9194287c724 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab0ccafcfe4e4d62b948928c6dd7b2d5 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0adaa9d1021344eeb717940a9ca3f5b8 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3649809672244eafa86f2b5ebe4cfb11 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9956eafee694f4e85ef92e0cc9f4023 : SUCCESS in 4m 09s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/a0cc49d9d70148c1a882f3d94b294c00 : FAILURE in 52m 51s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/df43fd6b07ab42bb942355b057eeb96f : FAILURE in 52m 45s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/a8c3f734716d4df1ba1a7ee79d4381c1 : FAILURE in 42m 44s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/562cdf76dd604d73b879921624641b40 : SUCCESS in 53m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3c4b4fe37384068ac71a8355265ede4 : SUCCESS in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a9ad59775e46659b0e07d6894a102d753f81cdbb","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2020-10-12 22:07:04.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"03cbd357833a83a7160fbb5be0e119fa429d8d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 23:12:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e3aa9a65de0405b8cc6d6d66440afd3 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af5b0e162d84d22b5619ea740d33bdf : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aec041470ccb45e1b6a9a5fce8c26c41 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8d96366c1db431a9ab208dae4a5a39b : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde50012e43b4f868e18b263447e0a49 : SUCCESS in 4m 49s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/e047e84958194b6983b496cba53a4edb : SUCCESS in 47m 02s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/9561288318534d26a60334705e189e5d : SUCCESS in 47m 12s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/bc88139496e94d69aec9a94b760fa43f : SUCCESS in 51m 07s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/75adb906f82e478684f92ad458dbfaec : SUCCESS in 48m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/643920267dc8437cbe8b4c21fcc5f61b : SUCCESS in 7m 34s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"04c83691c1a4f43f0572eabc896121d223467de3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-13 07:23:42.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"74059fe0c8bf86c2023495e54c07f1c31861fd21","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2020-10-13 16:09:35.000000000","message":"Patch Set 11: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":11},{"id":"f9939cc834036d77819637709fb547555d7c6d98","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-13 21:16:23.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"dc9634b848f28aa903db021e9fe160d5d21521ba","author":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"date":"2020-10-13 22:34:34.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"358b72da502235276839678e88c755814ba5ecd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 22:34:47.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"42b995c8ed4daea418d91b421081056e7978ba83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 23:30:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"f8750cf767cb6a0ff744562818e7bb2edcc25202","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 23:30:50.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e082885dc2c4389ac93c191e9c09671 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d97e056cb54ff7959bcffbc5b0d953 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/496be0df672049cc86668a81d961f427 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c99a2fb940247c385f41a5c9f9c36a5 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b5e7c64fd064695b749578cbd9a4c4f : SUCCESS in 4m 38s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/64c85829eee84dec82c0323b4d426b48 : SUCCESS in 48m 46s","accounts_in_message":[],"_revision_number":11},{"id":"65541a1fe42d35b9db48d703b9df46646e874380","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 23:32:14.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/318ca3f81a6b4f2892cead74455b8b08 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d4ec93ae55256e8ccf04eac45c5364c694a34759","revisions":{"a0370f992ca4b1b16a23b8d9e798b5efc972189f":{"kind":"REWORK","_number":1,"created":"2020-09-04 06:48:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/749874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/74/749874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/74/749874/1"}}},"commit":{"parents":[{"commit":"585bd14c70415b63265321cc7728b4e8c50702a7","subject":"Remove neutron-fwaas from the jobs\u0027 required project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/585bd14c70415b63265321cc7728b4e8c50702a7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:15.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:38.000000000","tz":180},"subject":"Add options to choose endpoints and SSL verification","message":"Add options to choose endpoints and SSL verification\n\nIn case of usage of self-signed SSL certificates for public endpoints,\nyou will get stuck with senlin usage. So in case of such scenario it\u0027s\nessential to either choose endpoint type or disable SSL verification.\nBoth of these options are implemented in most services and are expected.\n\nChange-Id: Ifa475ff146af8b49a762218ff244f0ff0fee9ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/a0370f992ca4b1b16a23b8d9e798b5efc972189f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/a0370f992ca4b1b16a23b8d9e798b5efc972189f"}]},"branch":"refs/heads/master"},"9e96739531b4f270d6bed8df7169a81698123e2a":{"kind":"REWORK","_number":2,"created":"2020-09-04 07:03:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/749874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/74/749874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/74/749874/2"}}},"commit":{"parents":[{"commit":"585bd14c70415b63265321cc7728b4e8c50702a7","subject":"Remove neutron-fwaas from the jobs\u0027 required project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/585bd14c70415b63265321cc7728b4e8c50702a7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:15.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 07:03:34.000000000","tz":180},"subject":"Add options to choose endpoints and SSL verification","message":"Add options to choose endpoints and SSL verification\n\nIn case of usage of self-signed SSL certificates for public endpoints,\nyou will get stuck with senlin usage. So in case of such scenario it\u0027s\nessential to either choose endpoint type or disable SSL verification.\nBoth of these options are implemented in most services and are expected.\n\nChange-Id: Ifa475ff146af8b49a762218ff244f0ff0fee9ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/9e96739531b4f270d6bed8df7169a81698123e2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/9e96739531b4f270d6bed8df7169a81698123e2a"}]},"branch":"refs/heads/master"},"29ea4066ca1965234945a4320e3bc3fb72bdc333":{"kind":"REWORK","_number":3,"created":"2020-09-04 08:38:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/749874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/74/749874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/74/749874/3"}}},"commit":{"parents":[{"commit":"585bd14c70415b63265321cc7728b4e8c50702a7","subject":"Remove neutron-fwaas from the jobs\u0027 required project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/585bd14c70415b63265321cc7728b4e8c50702a7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:15.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-04 08:38:15.000000000","tz":0},"subject":"Add options to choose endpoints and SSL verification","message":"Add options to choose endpoints and SSL verification\n\nIn case of usage of self-signed SSL certificates for public endpoints,\nyou will get stuck with senlin usage. So in case of such scenario it\u0027s\nessential to either choose endpoint type or disable SSL verification.\nBoth of these options are implemented in most services and are expected.\n\nChange-Id: Ifa475ff146af8b49a762218ff244f0ff0fee9ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/29ea4066ca1965234945a4320e3bc3fb72bdc333"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/29ea4066ca1965234945a4320e3bc3fb72bdc333"}]},"branch":"refs/heads/master"},"0981b9abe73afa57bfe42dfb9dd5e03da4604e5c":{"kind":"REWORK","_number":4,"created":"2020-09-04 09:42:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/749874/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/74/749874/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/74/749874/4"}}},"commit":{"parents":[{"commit":"585bd14c70415b63265321cc7728b4e8c50702a7","subject":"Remove neutron-fwaas from the jobs\u0027 required project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/585bd14c70415b63265321cc7728b4e8c50702a7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:15.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-04 09:42:09.000000000","tz":0},"subject":"Add options to choose endpoints and SSL verification","message":"Add options to choose endpoints and SSL verification\n\nIn case of usage of self-signed SSL certificates for public endpoints,\nyou will get stuck with senlin usage. So in case of such scenario it\u0027s\nessential to either choose endpoint type or disable SSL verification.\nBoth of these options are implemented in most services and are expected.\n\nChange-Id: Ifa475ff146af8b49a762218ff244f0ff0fee9ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/0981b9abe73afa57bfe42dfb9dd5e03da4604e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/0981b9abe73afa57bfe42dfb9dd5e03da4604e5c"}]},"branch":"refs/heads/master"},"2afeeffe61d0a644fa0bf956604fad5ce9cac856":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-14 12:56:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/749874/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/74/749874/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/74/749874/5"}}},"commit":{"parents":[{"commit":"907a9250e7f3125ed0f5a0fe38b965d5fc4c07f4","subject":"Merge \"Fix Flake F811 for microversioned methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/907a9250e7f3125ed0f5a0fe38b965d5fc4c07f4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:15.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-14 12:56:25.000000000","tz":0},"subject":"Add options to choose endpoints and SSL verification","message":"Add options to choose endpoints and SSL verification\n\nIn case of usage of self-signed SSL certificates for public endpoints,\nyou will get stuck with senlin usage. So in case of such scenario it\u0027s\nessential to either choose endpoint type or disable SSL verification.\nBoth of these options are implemented in most services and are expected.\n\nChange-Id: Ifa475ff146af8b49a762218ff244f0ff0fee9ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/2afeeffe61d0a644fa0bf956604fad5ce9cac856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/2afeeffe61d0a644fa0bf956604fad5ce9cac856"}]},"branch":"refs/heads/master"},"20edfd6ba846c3b34efa31d799762a25264e7754":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-09-16 13:46:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/749874/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/74/749874/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/74/749874/6"}}},"commit":{"parents":[{"commit":"aca5b42d81bf3ec422ea6452027f3527c3b5443b","subject":"Pass in correct port id parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/aca5b42d81bf3ec422ea6452027f3527c3b5443b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:15.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-16 13:46:25.000000000","tz":0},"subject":"Add options to choose endpoints and SSL verification","message":"Add options to choose endpoints and SSL verification\n\nIn case of usage of self-signed SSL certificates for public endpoints,\nyou will get stuck with senlin usage. So in case of such scenario it\u0027s\nessential to either choose endpoint type or disable SSL verification.\nBoth of these options are implemented in most services and are expected.\n\nChange-Id: Ifa475ff146af8b49a762218ff244f0ff0fee9ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/20edfd6ba846c3b34efa31d799762a25264e7754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/20edfd6ba846c3b34efa31d799762a25264e7754"}]},"branch":"refs/heads/master"},"f068efee9a48cdd0e54953c56002b385a4468842":{"kind":"REWORK","_number":7,"created":"2020-10-01 07:24:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/749874/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/74/749874/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/74/749874/7"}}},"commit":{"parents":[{"commit":"6ff278642ec8ff62bc6f943f7a0b47c485c23e41","subject":"Release notes for Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/6ff278642ec8ff62bc6f943f7a0b47c485c23e41"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:15.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-01 07:24:33.000000000","tz":180},"subject":"Add options to choose endpoints and SSL verification","message":"Add options to choose endpoints and SSL verification\n\nIn case of usage of self-signed SSL certificates for public endpoints,\nyou will get stuck with senlin usage. So in case of such scenario it\u0027s\nessential to either choose endpoint type or disable SSL verification.\nBoth of these options are implemented in most services and are expected.\n\nChange-Id: Ifa475ff146af8b49a762218ff244f0ff0fee9ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/f068efee9a48cdd0e54953c56002b385a4468842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/f068efee9a48cdd0e54953c56002b385a4468842"}]},"branch":"refs/heads/master"},"f1d99fd5317a5a4ec2b0f20d87f53e089250ee0e":{"kind":"REWORK","_number":8,"created":"2020-10-01 17:28:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/749874/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/74/749874/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/74/749874/8"}}},"commit":{"parents":[{"commit":"6ff278642ec8ff62bc6f943f7a0b47c485c23e41","subject":"Release notes for Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/6ff278642ec8ff62bc6f943f7a0b47c485c23e41"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:15.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-01 17:27:53.000000000","tz":180},"subject":"Add options to choose endpoints and SSL verification","message":"Add options to choose endpoints and SSL verification\n\nIn case of usage of self-signed SSL certificates for public endpoints,\nyou will get stuck with senlin usage. So in case of such scenario it\u0027s\nessential to either choose endpoint type or disable SSL verification.\nBoth of these options are implemented in most services and are expected.\n\nChange-Id: Ifa475ff146af8b49a762218ff244f0ff0fee9ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/f1d99fd5317a5a4ec2b0f20d87f53e089250ee0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/f1d99fd5317a5a4ec2b0f20d87f53e089250ee0e"}]},"branch":"refs/heads/master"},"8499a1471466cd85fd119f36a7feddee6b35e7b7":{"kind":"REWORK","_number":9,"created":"2020-10-02 08:53:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/749874/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/74/749874/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/74/749874/9"}}},"commit":{"parents":[{"commit":"6ff278642ec8ff62bc6f943f7a0b47c485c23e41","subject":"Release notes for Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/6ff278642ec8ff62bc6f943f7a0b47c485c23e41"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:15.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-02 08:53:04.000000000","tz":180},"subject":"Add options to choose endpoints and SSL verification","message":"Add options to choose endpoints and SSL verification\n\nIn case of usage of self-signed SSL certificates for public endpoints,\nyou will get stuck with senlin usage. So in case of such scenario it\u0027s\nessential to either choose endpoint type or disable SSL verification.\nBoth of these options are implemented in most services and are expected.\n\nChange-Id: Ifa475ff146af8b49a762218ff244f0ff0fee9ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/8499a1471466cd85fd119f36a7feddee6b35e7b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/8499a1471466cd85fd119f36a7feddee6b35e7b7"}]},"branch":"refs/heads/master"},"00a13bd0a74e0f44b4f66ccb78e80c8d9d329f2b":{"kind":"REWORK","_number":10,"created":"2020-10-02 18:16:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/749874/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/74/749874/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/74/749874/10"}}},"commit":{"parents":[{"commit":"6ff278642ec8ff62bc6f943f7a0b47c485c23e41","subject":"Release notes for Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/6ff278642ec8ff62bc6f943f7a0b47c485c23e41"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:15.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-02 18:16:03.000000000","tz":180},"subject":"Add options to choose endpoints and SSL verification","message":"Add options to choose endpoints and SSL verification\n\nIn case of usage of self-signed SSL certificates for public endpoints,\nyou will get stuck with senlin usage. So in case of such scenario it\u0027s\nessential to either choose endpoint type or disable SSL verification.\nBoth of these options are implemented in most services and are expected.\n\nChange-Id: Ifa475ff146af8b49a762218ff244f0ff0fee9ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/00a13bd0a74e0f44b4f66ccb78e80c8d9d329f2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/00a13bd0a74e0f44b4f66ccb78e80c8d9d329f2b"}]},"branch":"refs/heads/master"},"d4ec93ae55256e8ccf04eac45c5364c694a34759":{"kind":"REWORK","_number":11,"created":"2020-10-12 22:07:04.000000000","uploader":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"ref":"refs/changes/74/749874/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/74/749874/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/74/749874/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/74/749874/11"}}},"commit":{"parents":[{"commit":"6ff278642ec8ff62bc6f943f7a0b47c485c23e41","subject":"Release notes for Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/6ff278642ec8ff62bc6f943f7a0b47c485c23e41"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 06:48:15.000000000","tz":180},"committer":{"name":"Duc Truong","email":"dtruong@blizzard.com","date":"2020-10-12 22:07:04.000000000","tz":0},"subject":"Add options to choose endpoints and SSL verification","message":"Add options to choose endpoints and SSL verification\n\nIn case of usage of self-signed SSL certificates for public endpoints,\nyou will get stuck with senlin usage. So in case of such scenario it\u0027s\nessential to either choose endpoint type or disable SSL verification.\nBoth of these options are implemented in most services and are expected.\n\nChange-Id: Ifa475ff146af8b49a762218ff244f0ff0fee9ac0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d4ec93ae55256e8ccf04eac45c5364c694a34759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d4ec93ae55256e8ccf04eac45c5364c694a34759"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
