)]}'
{"id":"openstack%2Fsenlin~823293","triplet_id":"openstack%2Fsenlin~stable%2Frocky~I8e99d7c76dff1ba594385b828fabdf5ab94b9b1f","project":"openstack/senlin","branch":"stable/rocky","topic":"remove-neutron-lbaas-stable/rocky","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-09-24 05:15:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8e99d7c76dff1ba594385b828fabdf5ab94b9b1f","subject":"Remove neutron-lbaas from the CI jobs requirements","status":"ABANDONED","created":"2022-01-03 13:35:32.000000000","updated":"2022-09-24 05:15:24.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":823287,"cherry_pick_of_patch_set":1,"meta_rev_id":"278d7973c591f4652c6de3dd170f092e002b27ea","_number":823293,"virtual_id_number":823293,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-03 14:35:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-03 14:35:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-07 07:11:04.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"f10326db337edce941726e170eddf74156a584b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-03 13:35:32.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"deb88a5f24606f52aa19faa6d2154098a92ccc81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 14:35:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a545d51df42d45838d741e910f4474b0 : FAILURE in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81943e42f06144f4a8350a1270392909 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b2434b1594447d1a135864276691060 : FAILURE in 3m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a22a6641c7db45bb9d2f8be327de97a7 : SUCCESS in 6m 16s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/91b56346b6604dc8bbf526345d2fcd20 : FAILURE in 33m 41s\n- senlin-dsvm-tempest-py35-api https://zuul.opendev.org/t/openstack/build/e53363bcdcd545efa0eb701ba688396c : SUCCESS in 50m 05s (non-voting)\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/c15831908eea41ffad698f1d2167ee65 : FAILURE in 37m 27s\n- senlin-dsvm-tempest-py35-functional https://zuul.opendev.org/t/openstack/build/878c0a3a0b1b416ba7c6cf4de3aa1a61 : SUCCESS in 37m 04s (non-voting)\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/14a107fcc57d4323b14fa92400510d5f : FAILURE in 36m 54s (non-voting)\n- senlin-dsvm-tempest-py35-integration https://zuul.opendev.org/t/openstack/build/80d0c8ec0d0d4e12a03b90b019c4c192 : FAILURE in 34m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a941fc32890f4f85bd73d877109a673c : FAILURE in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbbda132c0f64ac88e7650f02ac24dca : SUCCESS in 7m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf223054707b562384dd0f66c540dcb9232d81a6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-07 07:11:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c00117a7b1434b37f01c5532569030df78a4412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 08:09:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a8f9d89f69764b03a373ef0178cd73d1 : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a2c446bb6c4b48ba9722c2a293f424 : SUCCESS in 10m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2f6d5b3d1474009acd02a905e426abb : FAILURE in 3m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a12bef396bea40bb9ec9eb7a9c6674d2 : SUCCESS in 11m 46s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/8f70e24fa3bb42f59e7bd198659bc540 : FAILURE in 26m 45s\n- senlin-dsvm-tempest-py35-api https://zuul.opendev.org/t/openstack/build/52ffc6d9aa7342689d9b1e9405596b7f : SUCCESS in 56m 40s (non-voting)\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/dc147acd8a094bbfa553cfe7a4f63ce5 : FAILURE in 35m 46s\n- senlin-dsvm-tempest-py35-functional https://zuul.opendev.org/t/openstack/build/64ebacd830324d6998d24a9cffa09f42 : SUCCESS in 44m 49s (non-voting)\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/c1e8c746cb604d5aa623ee0c54b6fa95 : FAILURE in 33m 06s (non-voting)\n- senlin-dsvm-tempest-py35-integration https://zuul.opendev.org/t/openstack/build/1780bf2435a749a190e85c00289f6972 : FAILURE in 34m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0980e83d4afc44bca2f3ffcd8e6d0238 : FAILURE in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4614e51328bc496894c3e0c881619fd4 : SUCCESS in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"278d7973c591f4652c6de3dd170f092e002b27ea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:15:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2a85d50aec25a4f021dbf1aab936e997339ce7d","revisions":{"b2a85d50aec25a4f021dbf1aab936e997339ce7d":{"kind":"REWORK","_number":1,"created":"2022-01-03 13:35:32.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/93/823293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/93/823293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/93/823293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/93/823293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/93/823293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/93/823293/1"}}},"commit":{"parents":[{"commit":"4279338f55009dfd7d8ddc5bf444dc35a1b41104","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/4279338f55009dfd7d8ddc5bf444dc35a1b41104"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-03 13:34:41.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-03 13:35:32.000000000","tz":0},"subject":"Remove neutron-lbaas from the CI jobs requirements","message":"Remove neutron-lbaas from the CI jobs requirements\n\nNeutron-lbaas was retired some time ago [1] so there\u0027s no need\nto keep it in the requirements of the CI jobs.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006158.html\n\nChange-Id: I8e99d7c76dff1ba594385b828fabdf5ab94b9b1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/b2a85d50aec25a4f021dbf1aab936e997339ce7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/b2a85d50aec25a4f021dbf1aab936e997339ce7d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
