)]}'
{"id":"openstack%2Fsenlin~823294","triplet_id":"openstack%2Fsenlin~stable%2Fqueens~I8e99d7c76dff1ba594385b828fabdf5ab94b9b1f","project":"openstack/senlin","branch":"stable/queens","topic":"remove-neutron-lbaas-stable/rocky-stable/queens","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-09-24 05:15:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8e99d7c76dff1ba594385b828fabdf5ab94b9b1f","subject":"Remove neutron-lbaas from the CI jobs requirements","status":"ABANDONED","created":"2022-01-03 13:35:43.000000000","updated":"2022-09-24 05:15:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":823293,"cherry_pick_of_patch_set":1,"meta_rev_id":"873922b4c6463f40c62022406139c256654106bb","_number":823294,"virtual_id_number":823294,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-11 10:31:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-03 13:36:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1bdcfdfd6a660e3ec49b06db9353201cb5ef161","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-03 13:35:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"3957c1430591b34c6d4d1964812c047667b8e027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 13:36:42.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/senlin on branch stable/queens.  The error was:\n\n  Unknown projects: openstack/neutron-lbaas\n\nThe error appears in the following job stanza:\n\n  job:\n      name: senlin-dsvm-tempest-py27-functional\n      parent: legacy-dsvm-base\n      run: playbooks/legacy/senlin-dsvm-tempest-py27-functional/run.yaml\n      post-run: playbooks/legacy/senlin-dsvm-tempest-py27-functional/post.yaml\n      timeout: 7800\n      required-projects:\n        - openstack/devstack-gate\n        - openstack/heat\n        - openstack/neutron-lbaas\n        - openstack/python-zaqarclient\n        - openstack/senlin\n        - openstack/senlin-tempest-plugin\n        - openstack/tempest\n        - openstack/zaqar\n        - openstack/zaqar-ui\n      irrelevant-files:\n        - ^.*\\.rst$\n        - ^api-ref/.*$\n        - ^doc/.*$\n        - ^senlin/tests/unit/.*$\n        - ^releasenotes/.*$\n\n  in \"openstack/senlin/.zuul.yaml@stable/queens\", line 54, column 3","accounts_in_message":[],"_revision_number":1},{"id":"4c508264b7a468b6fe50a9509c2c4ca31452fc5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-11 09:58:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"89ebe3198e8640596903ee7c7f0a7f3dec74a85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 10:31:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ce66ce1e0349463997503702e557d530 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b146c0f3f1934333ad92464e39c4808c : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/303d9a06ca8441eea75d64106109f5ea : FAILURE in 3m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/699753b4b524450ba5cf2a2ec638f83b : SUCCESS in 5m 27s\n- senlin-dsvm-tempest-py27-api https://zuul.opendev.org/t/openstack/build/06621f459d5e4273a34551b14954dd5a : FAILURE in 25m 04s (non-voting)\n- senlin-dsvm-tempest-py35-api https://zuul.opendev.org/t/openstack/build/e63b7f0aeca7432ea2da2f2cef0849fa : FAILURE in 25m 39s (non-voting)\n- senlin-dsvm-tempest-py27-functional https://zuul.opendev.org/t/openstack/build/745b16e6f18447269ee8b44109ac2a23 : FAILURE in 24m 34s\n- senlin-dsvm-tempest-py35-functional https://zuul.opendev.org/t/openstack/build/609e1441b066456ab900546ef6024175 : FAILURE in 29m 02s (non-voting)\n- senlin-dsvm-tempest-py27-integration https://zuul.opendev.org/t/openstack/build/a9f469f68e124677aa39a28320441321 : FAILURE in 9m 39s (non-voting)\n- senlin-dsvm-tempest-py35-integration https://zuul.opendev.org/t/openstack/build/fb0fd1e8c30448b5a8db9b66f7899376 : FAILURE in 15m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64eb75efb2014fea9f06b41500ba8e0c : FAILURE in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"873922b4c6463f40c62022406139c256654106bb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:15:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d20fff89b09d585c7924b71aa8ff4e5a6cf8d3f1","revisions":{"d7bf0736c764fdb4066c48ba4dd103e9059ea789":{"kind":"REWORK","_number":1,"created":"2022-01-03 13:35:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/94/823294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/94/823294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/94/823294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/94/823294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/94/823294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/94/823294/1"}}},"commit":{"parents":[{"commit":"1adbf9c4db0e916e5d3bc9e58145316802133dff","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/1adbf9c4db0e916e5d3bc9e58145316802133dff"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-03 13:34:41.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-03 13:35:43.000000000","tz":0},"subject":"Remove neutron-lbaas from the CI jobs requirements","message":"Remove neutron-lbaas from the CI jobs requirements\n\nNeutron-lbaas was retired some time ago [1] so there\u0027s no need\nto keep it in the requirements of the CI jobs.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006158.html\n\nChange-Id: I8e99d7c76dff1ba594385b828fabdf5ab94b9b1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d7bf0736c764fdb4066c48ba4dd103e9059ea789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d7bf0736c764fdb4066c48ba4dd103e9059ea789"}]},"branch":"refs/heads/stable/queens"},"d20fff89b09d585c7924b71aa8ff4e5a6cf8d3f1":{"kind":"REWORK","_number":2,"created":"2022-04-11 09:58:52.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/94/823294/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/94/823294/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/94/823294/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/94/823294/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/94/823294/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/94/823294/2"}}},"commit":{"parents":[{"commit":"1adbf9c4db0e916e5d3bc9e58145316802133dff","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/1adbf9c4db0e916e5d3bc9e58145316802133dff"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-03 13:34:41.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-11 09:58:44.000000000","tz":0},"subject":"Remove neutron-lbaas from the CI jobs requirements","message":"Remove neutron-lbaas from the CI jobs requirements\n\nNeutron-lbaas was retired some time ago [1] so there\u0027s no need\nto keep it in the requirements of the CI jobs.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006158.html\n\nChange-Id: I8e99d7c76dff1ba594385b828fabdf5ab94b9b1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d20fff89b09d585c7924b71aa8ff4e5a6cf8d3f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/d20fff89b09d585c7924b71aa8ff4e5a6cf8d3f1"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
