)]}'
{"id":"openstack%2Fsenlin~868666","triplet_id":"openstack%2Fsenlin~master~I39f14fe35e15a1045f6ca22ad2740b0d1d38bdb1","project":"openstack/senlin","branch":"master","attention_set":{},"removed_from_attention_set":{"33881":{"account":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"last_update":"2023-01-09 16:35:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I39f14fe35e15a1045f6ca22ad2740b0d1d38bdb1","subject":"Fix tox.ini for tox4","status":"ABANDONED","created":"2022-12-30 02:45:48.000000000","updated":"2023-01-09 16:35:35.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8fd4f492b5fae7d50946822ac5d21cc0367a7852","_number":868666,"virtual_id_number":868666,"owner":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-30 03:38:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-30 03:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-31 20:09:00.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"84db202a7e0eba4d83de59e9130a06b6ed3a6b28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-12-30 02:45:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d10ae9a460d6df644f258d3189ab18af6e3f09e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-12-30 02:47:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a65efb1635f9a909955769963483922be5e4a332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 03:38:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d958f83fe36402e86c6cdbd34189705\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/07115417b07f4b0b94bdf3213eecb1b3 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd40b232cc714e4cb2f4dae773fda9a3 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bf3590a82414b75958cb92e9fceabe2 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5221723feac3445681e9bf3fbf6eba4c : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27770b5f9fdb48d0a85f266abf5ab953 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3122d34c564343ec8c18558f9c4eade5 : SUCCESS in 5m 40s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/cf781b241d4d4dbfa04be5cf2d488eb3 : SUCCESS in 33m 07s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/415e86aa2bdf4d59a5bbeea2e30c2126 : SUCCESS in 48m 06s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/208e9e83924a4151b6e77676e3c8900b : SUCCESS in 28m 42s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/abe09a6736754c5bbc1c534f80006a2d : FAILURE in 41m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17c6bd854704470e98a2350889c1d755 : FAILURE in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ebdbcb031cdf6cf58c40496c75d7d3d123b3e147","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-12-31 20:09:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abd8c7b957104541e8eceac5fa9bcd08702e4e74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 20:39:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/daa95a73f47740999534583691e4f963\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/93f34e8955a546a7800706f9b689569a : FAILURE in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f12c2e0566e42f4845f15cb620aaea4 : FAILURE in 2m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d9ca121560641c6a5ab65f6baa13fed : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea5d22d1d17d4a8e853269d2f6f9c30e : FAILURE in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1aadd5d2c14c3d82d0ca544fad2ac1 : FAILURE in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8cbc4c24b2e4485ba51d4001a225e7d : SUCCESS in 3m 44s\n- senlin-dsvm-tempest-py3-api https://zuul.opendev.org/t/openstack/build/ee7360598b01481eb9d766ad88339e1e : SUCCESS in 29m 03s\n- senlin-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/3ee28f68c36343c08da2d751a6504725 : SUCCESS in 29m 08s\n- senlin-dsvm-tempest-py3-functional https://zuul.opendev.org/t/openstack/build/2cb23a81610b4f919d6cecdc29f5ca68 : SUCCESS in 24m 43s\n- senlin-dsvm-tempest-py3-integration https://zuul.opendev.org/t/openstack/build/f2fa3ae10ef94f9a9088217767a4d0cc : FAILURE in 16m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7edfcf35e71409d91a8695268e5e875 : FAILURE in 2m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f25d8ae3d90c0fe7f400a18c3e0fe82b1b913136","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-06 00:07:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8fd4f492b5fae7d50946822ac5d21cc0367a7852","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-01-09 16:35:35.000000000","message":"Abandoned\n\nI am abandoning this patch as another tox fix merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f5e03f4c3b1e379aa6332dd204139b9ad41a101f","revisions":{"448fd8d50c031f23eda2d4f88aaea6a6068621cc":{"kind":"REWORK","_number":1,"created":"2022-12-30 02:45:48.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/66/868666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/66/868666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/66/868666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/66/868666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/66/868666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/66/868666/1"}}},"commit":{"parents":[{"commit":"0743b6ec7b31831430422bc5513e05820a326a0c","subject":"Add the lost component when in installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/0743b6ec7b31831430422bc5513e05820a326a0c"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-12-30 02:42:31.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-12-30 02:42:31.000000000","tz":480},"subject":"Fix tox.ini for tox4","message":"Fix tox.ini for tox4\n\nChange-Id: I39f14fe35e15a1045f6ca22ad2740b0d1d38bdb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/448fd8d50c031f23eda2d4f88aaea6a6068621cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/448fd8d50c031f23eda2d4f88aaea6a6068621cc"}]},"branch":"refs/heads/master"},"f5e03f4c3b1e379aa6332dd204139b9ad41a101f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-30 02:47:31.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/66/868666/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/senlin","ref":"refs/changes/66/868666/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/senlin refs/changes/66/868666/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/senlin refs/changes/66/868666/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/senlin refs/changes/66/868666/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/senlin refs/changes/66/868666/2"}}},"commit":{"parents":[{"commit":"0743b6ec7b31831430422bc5513e05820a326a0c","subject":"Add the lost component when in installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/0743b6ec7b31831430422bc5513e05820a326a0c"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-12-30 02:42:31.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-12-30 02:47:31.000000000","tz":0},"subject":"Fix tox.ini for tox4","message":"Fix tox.ini for tox4\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943/\n\nChange-Id: I39f14fe35e15a1045f6ca22ad2740b0d1d38bdb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/f5e03f4c3b1e379aa6332dd204139b9ad41a101f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/senlin/commit/f5e03f4c3b1e379aa6332dd204139b9ad41a101f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
