)]}'
{"id":"openstack%2Fshade~654235","triplet_id":"openstack%2Fshade~master~I2cf0463108cf0bea37a221ee1dd74252d13f41c4","project":"openstack/shade","branch":"master","topic":"opendev-gerrit","hashtags":[],"change_id":"I2cf0463108cf0bea37a221ee1dd74252d13f41c4","subject":"Update for opendev","status":"MERGED","created":"2019-04-21 13:00:52.000000000","updated":"2019-04-21 14:47:16.000000000","submitted":"2019-04-21 14:47:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654235-1555858036693-dfe5b5ea","meta_rev_id":"14583f3265c315b7d1d55b409abf999b91e25453","_number":654235,"virtual_id_number":654235,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-04-21 14:47:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-04-21 13:52:47.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-04-21 13:52:47.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-21 14:47:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6760be0d57fadd1ec57df82e4940269407d6b34c","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-04-21 13:00:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"899e0a7003e83a2fcdcb7e77cb81df9c71a67435","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 13:50:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/35/654235/1/check/test-release-openstack/6f10ccd/ : SUCCESS in 1m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/654235/1/check/openstack-tox-lower-constraints/02b0c28/ : SUCCESS in 2m 34s\n- openstack-tox-pep8 http://logs.openstack.org/35/654235/1/check/openstack-tox-pep8/1d4e0aa/ : SUCCESS in 2m 06s\n- openstack-tox-py27 http://logs.openstack.org/35/654235/1/check/openstack-tox-py27/7920ddd/ : SUCCESS in 2m 28s\n- openstack-tox-py36 http://logs.openstack.org/35/654235/1/check/openstack-tox-py36/1c937cf/ : SUCCESS in 2m 26s\n- openstack-tox-docs http://logs.openstack.org/35/654235/1/check/openstack-tox-docs/bc1c1c8/html/ : SUCCESS in 2m 20s\n- build-openstack-releasenotes http://logs.openstack.org/35/654235/1/check/build-openstack-releasenotes/076f2a9/html/ : SUCCESS in 2m 23s\n- shade-functional-devstack-tips http://logs.openstack.org/35/654235/1/check/shade-functional-devstack-tips/dcce495/ : SUCCESS in 44m 58s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/35/654235/1/check/shade-functional-devstack-tips-python3/6abcab0/ : SUCCESS in 38m 47s\n- shade-tox-py27-tips http://logs.openstack.org/35/654235/1/check/shade-tox-py27-tips/879ba12/ : SUCCESS in 3m 09s\n- shade-tox-py35-tips http://logs.openstack.org/35/654235/1/check/shade-tox-py35-tips/2430a7c/ : SUCCESS in 3m 02s\n- bifrost-integration-tinyipa http://logs.openstack.org/35/654235/1/check/bifrost-integration-tinyipa/93c121c/ : FAILURE in 3m 59s (non-voting)\n- bifrost-integration-tinyipa-opensuse-150 http://logs.openstack.org/35/654235/1/check/bifrost-integration-tinyipa-opensuse-150/b832f8d/ : FAILURE in 9m 25s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack http://logs.openstack.org/35/654235/1/check/shade-ansible-stable-2.5-functional-devstack/5d7d5e7/ : SUCCESS in 30m 47s (non-voting)\n- shade-ansible-functional-devstack http://logs.openstack.org/35/654235/1/check/shade-ansible-functional-devstack/836d40b/ : SUCCESS in 33m 29s\n- shade-functional-devstack http://logs.openstack.org/35/654235/1/check/shade-functional-devstack/5a5969b/ : SUCCESS in 42m 30s\n- shade-functional-devstack-magnum http://logs.openstack.org/35/654235/1/check/shade-functional-devstack-magnum/e9b8e57/ : SUCCESS in 33m 18s (non-voting)\n- shade-functional-devstack-python3 http://logs.openstack.org/35/654235/1/check/shade-functional-devstack-python3/bed7f47/ : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1d9664d3fc5731ece568f077022768b7f5802269","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-04-21 13:52:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6bc5497be698b4a6c0171714cf9bf65788262fbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 13:52:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac469a877a2a63fe175b49a5fb59cbe43e7a99aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 14:47:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d60b58fb3cb64d8a71dd25d1bd38ac915fb4a128","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 14:47:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/35/654235/1/gate/test-release-openstack/981aefe/ : SUCCESS in 1m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/654235/1/gate/openstack-tox-lower-constraints/12a3cb1/ : SUCCESS in 2m 34s\n- openstack-tox-pep8 http://logs.openstack.org/35/654235/1/gate/openstack-tox-pep8/3f3b8bc/ : SUCCESS in 2m 16s\n- openstack-tox-py27 http://logs.openstack.org/35/654235/1/gate/openstack-tox-py27/2364fd5/ : SUCCESS in 2m 33s\n- openstack-tox-py36 http://logs.openstack.org/35/654235/1/gate/openstack-tox-py36/4b5467e/ : SUCCESS in 2m 27s\n- openstack-tox-docs http://logs.openstack.org/35/654235/1/gate/openstack-tox-docs/0444653/html/ : SUCCESS in 2m 25s\n- build-openstack-releasenotes http://logs.openstack.org/35/654235/1/gate/build-openstack-releasenotes/7d3430d/html/ : SUCCESS in 2m 21s\n- shade-functional-devstack-tips http://logs.openstack.org/35/654235/1/gate/shade-functional-devstack-tips/593825a/ : SUCCESS in 46m 53s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/35/654235/1/gate/shade-functional-devstack-tips-python3/6edd077/ : SUCCESS in 49m 07s\n- shade-tox-py27-tips http://logs.openstack.org/35/654235/1/gate/shade-tox-py27-tips/9cce72f/ : SUCCESS in 2m 57s\n- shade-tox-py35-tips http://logs.openstack.org/35/654235/1/gate/shade-tox-py35-tips/4c156cd/ : SUCCESS in 3m 49s\n- shade-ansible-functional-devstack http://logs.openstack.org/35/654235/1/gate/shade-ansible-functional-devstack/ff27f14/ : SUCCESS in 37m 35s\n- shade-functional-devstack http://logs.openstack.org/35/654235/1/gate/shade-functional-devstack/abfe442/ : SUCCESS in 43m 06s\n- shade-functional-devstack-python3 http://logs.openstack.org/35/654235/1/gate/shade-functional-devstack-python3/69340f2/ : SUCCESS in 44m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b47fb2af55d76dc171d1e3a2960ca1c8060a85e4","revisions":{"b47fb2af55d76dc171d1e3a2960ca1c8060a85e4":{"kind":"REWORK","_number":1,"created":"2019-04-21 13:00:52.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/35/654235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/35/654235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/35/654235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/35/654235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/35/654235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/35/654235/1"}}},"commit":{"parents":[{"commit":"a3ff9afbf02e006b4f348feb6e1e94bcff4decf7","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/a3ff9afbf02e006b4f348feb6e1e94bcff4decf7"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-04-21 13:00:31.000000000","tz":0},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-04-21 13:00:42.000000000","tz":0},"subject":"Update for opendev","message":"Update for opendev\n\nChange-Id: I2cf0463108cf0bea37a221ee1dd74252d13f41c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/b47fb2af55d76dc171d1e3a2960ca1c8060a85e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/b47fb2af55d76dc171d1e3a2960ca1c8060a85e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
