)]}'
{"id":"openstack%2Fshade~663922","triplet_id":"openstack%2Fshade~master~I170277293eb4ed1a12cf62bcf863707cb3d918eb","project":"openstack/shade","branch":"master","topic":"story_2005828/task_33585","hashtags":[],"change_id":"I170277293eb4ed1a12cf62bcf863707cb3d918eb","subject":"URL encode swift objects endpoints","status":"MERGED","created":"2019-06-07 12:35:05.000000000","updated":"2019-07-24 12:36:40.000000000","submitted":"2019-07-24 12:36:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663922-1563971800816-02757940","meta_rev_id":"46c7a0eaf8a873d48a6039f8ab762ee92dda7b8e","_number":663922,"virtual_id_number":663922,"owner":{"_account_id":8674,"name":"Yves-Gwenael Bourhis","email":"ygbourhis@gmail.com","username":"yves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":986,"name":"Rosario Di Somma","email":"rosario.disomma@gmail.com","username":"mr-rods"},{"value":0,"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},{"value":0,"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":2,"date":"2019-07-24 12:36:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-07-24 11:42:15.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":986,"name":"Rosario Di Somma","email":"rosario.disomma@gmail.com","username":"mr-rods"},{"value":0,"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},{"value":0,"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-07-24 11:42:15.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2019-06-13 08:42:19.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2019-06-13 08:42:11.000000000","_account_id":986,"name":"Rosario Di Somma","email":"rosario.disomma@gmail.com","username":"mr-rods"},{"value":0,"date":"2019-06-13 08:46:38.000000000","_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},{"value":0,"date":"2019-06-13 08:42:00.000000000","_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":986,"name":"Rosario Di Somma","email":"rosario.disomma@gmail.com","username":"mr-rods"},{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 08:42:00.000000000","updated_by":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"reviewer":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"state":"REVIEWER"},{"updated":"2019-06-13 08:42:11.000000000","updated_by":{"_account_id":986,"name":"Rosario Di Somma","email":"rosario.disomma@gmail.com","username":"mr-rods"},"reviewer":{"_account_id":986,"name":"Rosario Di Somma","email":"rosario.disomma@gmail.com","username":"mr-rods"},"state":"REVIEWER"},{"updated":"2019-06-13 08:42:19.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-06-13 08:46:38.000000000","updated_by":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"reviewer":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"state":"REVIEWER"},{"updated":"2019-07-24 11:42:15.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-24 12:36:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bc21a22564c315451ff9b517d42ff8a939895e2","author":{"_account_id":8674,"name":"Yves-Gwenael Bourhis","email":"ygbourhis@gmail.com","username":"yves"},"date":"2019-06-07 12:35:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdec46503af7b0b8f74e532456845480a85e9815","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:33:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/22/663922/1/check/openstack-tox-lower-constraints/7058f55/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/22/663922/1/check/openstack-tox-pep8/415ee56/ : SUCCESS in 2m 31s\n- openstack-tox-py27 http://logs.openstack.org/22/663922/1/check/openstack-tox-py27/dae7b7a/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/22/663922/1/check/openstack-tox-py36/949265b/ : SUCCESS in 2m 53s\n- openstack-tox-docs http://logs.openstack.org/22/663922/1/check/openstack-tox-docs/ae796ff/html/ : SUCCESS in 2m 36s\n- shade-functional-devstack-tips http://logs.openstack.org/22/663922/1/check/shade-functional-devstack-tips/4b88c3a/ : SUCCESS in 45m 57s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/22/663922/1/check/shade-functional-devstack-tips-python3/55cc468/ : SUCCESS in 47m 21s\n- shade-tox-py27-tips http://logs.openstack.org/22/663922/1/check/shade-tox-py27-tips/fafcb83/ : SUCCESS in 3m 50s\n- shade-tox-py35-tips http://logs.openstack.org/22/663922/1/check/shade-tox-py35-tips/bebb77b/ : SUCCESS in 2m 57s\n- bifrost-integration-tinyipa http://logs.openstack.org/22/663922/1/check/bifrost-integration-tinyipa/91f3cd9/ : FAILURE in 4m 37s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack http://logs.openstack.org/22/663922/1/check/shade-ansible-stable-2.5-functional-devstack/07f86ce/ : SUCCESS in 40m 43s (non-voting)\n- shade-ansible-functional-devstack http://logs.openstack.org/22/663922/1/check/shade-ansible-functional-devstack/ce7ab7f/ : SUCCESS in 40m 39s\n- shade-functional-devstack http://logs.openstack.org/22/663922/1/check/shade-functional-devstack/6c98561/ : SUCCESS in 50m 09s\n- shade-functional-devstack-magnum http://logs.openstack.org/22/663922/1/check/shade-functional-devstack-magnum/640d681/ : SUCCESS in 46m 58s (non-voting)\n- shade-functional-devstack-python3 http://logs.openstack.org/22/663922/1/check/shade-functional-devstack-python3/ecc634d/ : SUCCESS in 54m 08s","accounts_in_message":[],"_revision_number":1},{"id":"da2f3d39f1c8fa41712699fff4cb32c94304f8e4","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-24 11:42:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"500193e7fa8d19137f47ad15850ed0a346c981cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 11:42:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fed92ba7cb4aa013ed60e741e8dcd04e32016427","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 12:36:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/22/663922/1/gate/openstack-tox-lower-constraints/055570c/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/22/663922/1/gate/openstack-tox-pep8/e9f29a9/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/22/663922/1/gate/openstack-tox-py27/ac52170/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/22/663922/1/gate/openstack-tox-py36/7acb5ae/ : SUCCESS in 3m 36s\n- openstack-tox-py37 http://logs.openstack.org/22/663922/1/gate/openstack-tox-py37/b13c98b/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/22/663922/1/gate/openstack-tox-docs/4b3040f/html/ : SUCCESS in 2m 59s\n- shade-functional-devstack-tips http://logs.openstack.org/22/663922/1/gate/shade-functional-devstack-tips/792e5ff/ : SUCCESS in 48m 30s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/22/663922/1/gate/shade-functional-devstack-tips-python3/9b93d7a/ : SUCCESS in 48m 04s\n- shade-tox-py27-tips http://logs.openstack.org/22/663922/1/gate/shade-tox-py27-tips/e026880/ : SUCCESS in 4m 59s\n- shade-tox-py35-tips http://logs.openstack.org/22/663922/1/gate/shade-tox-py35-tips/1221211/ : SUCCESS in 4m 54s\n- shade-ansible-functional-devstack http://logs.openstack.org/22/663922/1/gate/shade-ansible-functional-devstack/fda8fab/ : SUCCESS in 36m 48s\n- shade-functional-devstack http://logs.openstack.org/22/663922/1/gate/shade-functional-devstack/831520f/ : SUCCESS in 43m 35s\n- shade-functional-devstack-python3 http://logs.openstack.org/22/663922/1/gate/shade-functional-devstack-python3/94ba415/ : SUCCESS in 44m 54s","accounts_in_message":[],"_revision_number":1},{"id":"304cec84f4d62287ed6a5dc5d59ab6e6624ca4c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 12:36:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e183ca4d23546bda67c53975cc0bc7651381621","revisions":{"8e183ca4d23546bda67c53975cc0bc7651381621":{"kind":"REWORK","_number":1,"created":"2019-06-07 12:35:05.000000000","uploader":{"_account_id":8674,"name":"Yves-Gwenael Bourhis","email":"ygbourhis@gmail.com","username":"yves"},"ref":"refs/changes/22/663922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/22/663922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/663922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/663922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/663922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/22/663922/1"}}},"commit":{"parents":[{"commit":"b47fb2af55d76dc171d1e3a2960ca1c8060a85e4","subject":"Update for opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/b47fb2af55d76dc171d1e3a2960ca1c8060a85e4"}]}],"author":{"name":"Yves-Gwenael Bourhis","email":"yves-gwenael.bourhis@orange.com","date":"2019-06-07 12:30:06.000000000","tz":120},"committer":{"name":"Yves-Gwenael Bourhis","email":"yves-gwenael.bourhis@orange.com","date":"2019-06-07 12:34:48.000000000","tz":120},"subject":"URL encode swift objects endpoints","message":"URL encode swift objects endpoints\n\nSwift containers and objects can contain any kind of characters.\nIf a container or name contains special characters, shade was failing to\naccess them.\nWe need to url encode containers and objects when creating the url to access\nthem.\n\nStory: 2005828\nTask: 33585\nChange-Id: I170277293eb4ed1a12cf62bcf863707cb3d918eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/8e183ca4d23546bda67c53975cc0bc7651381621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/8e183ca4d23546bda67c53975cc0bc7651381621"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
