)]}'
{"id":"openstack%2Fshade~668438","triplet_id":"openstack%2Fshade~master~Ieeeaee0ba81e55b2f75f96b20d305df5a5112188","project":"openstack/shade","branch":"master","hashtags":[],"change_id":"Ieeeaee0ba81e55b2f75f96b20d305df5a5112188","subject":"Mailing lists change openstack-dev to openstack-discuss","status":"MERGED","created":"2019-07-01 11:23:20.000000000","updated":"2019-07-05 16:27:21.000000000","submitted":"2019-07-05 16:27:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668438-1562344041248-49c2a3a2","meta_rev_id":"f85f63bc8c132fd8514c7a8a5d691bec6a1afded","_number":668438,"virtual_id_number":668438,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-07-05 16:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-07-05 16:07:02.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-07-05 16:07:02.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 16:07:02.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-05 16:27:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b13116bf97e2d233a2df73d0de7731bf2df55f5e","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2019-07-01 11:23:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d9fe19a883bc3e84ac28fb4c3b68e41beda50a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 11:39:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/38/668438/1/check/openstack-tox-pep8/83f601d/ : SUCCESS in 3m 04s\n- openstack-tox-docs http://logs.openstack.org/38/668438/1/check/openstack-tox-docs/fb85d3d/html/ : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3b66b47630ad71168213b56a930f439932a339ff","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-05 16:07:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1087aef578e1754550c8c9cd9a99e572757f9dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 16:07:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c1c787958b9b206d528e8e56693a9e2875cf206","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 16:27:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f85f63bc8c132fd8514c7a8a5d691bec6a1afded","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 16:27:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/38/668438/1/gate/openstack-tox-pep8/b8c62f8/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/38/668438/1/gate/openstack-tox-docs/c1e35d1/html/ : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0514bfb4e53e016de887bbc24c2aec7ddb31ecbb","revisions":{"0514bfb4e53e016de887bbc24c2aec7ddb31ecbb":{"kind":"REWORK","_number":1,"created":"2019-07-01 11:23:20.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/38/668438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/38/668438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/38/668438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/38/668438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/38/668438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/38/668438/1"}}},"commit":{"parents":[{"commit":"05de1cc854174ba0cea7da480328cc73a6a4ef93","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/05de1cc854174ba0cea7da480328cc73a6a4ef93"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2019-07-01 10:55:17.000000000","tz":480},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2019-07-01 11:23:13.000000000","tz":480},"subject":"Mailing lists change openstack-dev to openstack-discuss","message":"Mailing lists change openstack-dev to openstack-discuss\n\nChange-Id: Ieeeaee0ba81e55b2f75f96b20d305df5a5112188\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0514bfb4e53e016de887bbc24c2aec7ddb31ecbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0514bfb4e53e016de887bbc24c2aec7ddb31ecbb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
