)]}'
{"id":"openstack%2Fshade~669452","triplet_id":"openstack%2Fshade~master~I96c9f4dc64b252bac3c934936c7eebe270862c9e","project":"openstack/shade","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I96c9f4dc64b252bac3c934936c7eebe270862c9e","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 18:36:52.000000000","updated":"2019-07-09 12:43:46.000000000","submitted":"2019-07-09 12:43:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669452-1562676226728-7de5f3e4","meta_rev_id":"2fb7bdf09d8a0f6546e6864118113a20f3cca8ba","_number":669452,"virtual_id_number":669452,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-07-09 12:43:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-07-09 11:40:59.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-07-09 11:40:59.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 11:40:59.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-09 12:43:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47517eda2d0fbe4030c1915c33d9ebab9bf725df","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 18:36:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eece58fa4b5f4853f18feed3b18804cb971e523b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 19:35:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/52/669452/1/check/test-release-openstack/94d32c4/ : SUCCESS in 2m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/669452/1/check/openstack-tox-lower-constraints/2d908e8/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/52/669452/1/check/openstack-tox-pep8/273539e/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/52/669452/1/check/openstack-tox-py27/c80de14/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/52/669452/1/check/openstack-tox-py36/da07afa/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/52/669452/1/check/openstack-tox-py37/7685588/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/52/669452/1/check/openstack-tox-docs/3ebe78c/html/ : SUCCESS in 3m 13s\n- build-openstack-releasenotes http://logs.openstack.org/52/669452/1/check/build-openstack-releasenotes/48be962/html/ : SUCCESS in 3m 20s\n- shade-functional-devstack-tips http://logs.openstack.org/52/669452/1/check/shade-functional-devstack-tips/66453ce/ : SUCCESS in 46m 30s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/52/669452/1/check/shade-functional-devstack-tips-python3/7e2afc9/ : SUCCESS in 51m 04s\n- shade-tox-py27-tips http://logs.openstack.org/52/669452/1/check/shade-tox-py27-tips/e6ad967/ : SUCCESS in 3m 48s\n- shade-tox-py35-tips http://logs.openstack.org/52/669452/1/check/shade-tox-py35-tips/f6903d5/ : SUCCESS in 3m 29s\n- bifrost-integration-tinyipa http://logs.openstack.org/52/669452/1/check/bifrost-integration-tinyipa/02632ec/ : FAILURE in 5m 09s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack http://logs.openstack.org/52/669452/1/check/shade-ansible-stable-2.5-functional-devstack/1bde408/ : SUCCESS in 41m 19s (non-voting)\n- shade-ansible-functional-devstack http://logs.openstack.org/52/669452/1/check/shade-ansible-functional-devstack/785754e/ : SUCCESS in 38m 28s\n- shade-functional-devstack http://logs.openstack.org/52/669452/1/check/shade-functional-devstack/6abad53/ : SUCCESS in 43m 49s\n- shade-functional-devstack-magnum http://logs.openstack.org/52/669452/1/check/shade-functional-devstack-magnum/8835185/ : SUCCESS in 47m 20s (non-voting)\n- shade-functional-devstack-python3 http://logs.openstack.org/52/669452/1/check/shade-functional-devstack-python3/5fa7b2a/ : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e01015206fafa2c021ca2c3b54ea2201e5489de3","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-09 11:40:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19c372322d02972166d694025d561b44cce11cce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 11:41:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9ee106beb11d1ef8354a17bc6e0f5777ab17c77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 12:43:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2098f75773be06bd00a6b3b5f885d25aad155530","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 12:43:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/52/669452/1/gate/test-release-openstack/1a499f1/ : SUCCESS in 2m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/669452/1/gate/openstack-tox-lower-constraints/8cf1457/ : SUCCESS in 3m 13s\n- openstack-tox-pep8 http://logs.openstack.org/52/669452/1/gate/openstack-tox-pep8/25db631/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/52/669452/1/gate/openstack-tox-py27/e3c1287/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/52/669452/1/gate/openstack-tox-py36/de1bfed/ : SUCCESS in 3m 01s\n- openstack-tox-py37 http://logs.openstack.org/52/669452/1/gate/openstack-tox-py37/0d76e80/ : SUCCESS in 2m 51s\n- openstack-tox-docs http://logs.openstack.org/52/669452/1/gate/openstack-tox-docs/fd32521/html/ : SUCCESS in 2m 35s\n- build-openstack-releasenotes http://logs.openstack.org/52/669452/1/gate/build-openstack-releasenotes/5aa1da7/html/ : SUCCESS in 2m 57s\n- shade-functional-devstack-tips http://logs.openstack.org/52/669452/1/gate/shade-functional-devstack-tips/7c395f0/ : SUCCESS in 46m 56s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/52/669452/1/gate/shade-functional-devstack-tips-python3/c364c00/ : SUCCESS in 53m 02s\n- shade-tox-py27-tips http://logs.openstack.org/52/669452/1/gate/shade-tox-py27-tips/8f84f9c/ : SUCCESS in 4m 19s\n- shade-tox-py35-tips http://logs.openstack.org/52/669452/1/gate/shade-tox-py35-tips/9a2ad0d/ : SUCCESS in 4m 18s\n- shade-ansible-functional-devstack http://logs.openstack.org/52/669452/1/gate/shade-ansible-functional-devstack/394f6dd/ : SUCCESS in 34m 03s\n- shade-functional-devstack http://logs.openstack.org/52/669452/1/gate/shade-functional-devstack/4c5bda6/ : SUCCESS in 49m 59s\n- shade-functional-devstack-python3 http://logs.openstack.org/52/669452/1/gate/shade-functional-devstack-python3/2fa3187/ : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0639aa94f4061e85e350287cb488ea738587a47","revisions":{"b0639aa94f4061e85e350287cb488ea738587a47":{"kind":"REWORK","_number":1,"created":"2019-07-05 18:36:52.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/52/669452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/52/669452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/52/669452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/52/669452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/52/669452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/52/669452/1"}}},"commit":{"parents":[{"commit":"05de1cc854174ba0cea7da480328cc73a6a4ef93","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/05de1cc854174ba0cea7da480328cc73a6a4ef93"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:54:28.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 18:36:51.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I96c9f4dc64b252bac3c934936c7eebe270862c9e\nStory: #2005924\nTask: #34233\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/b0639aa94f4061e85e350287cb488ea738587a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/b0639aa94f4061e85e350287cb488ea738587a47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
