)]}'
{"id":"openstack%2Fshade~679736","triplet_id":"openstack%2Fshade~master~I1b0d95fd1eae8f0a901c788b2d6f4884c37eeec6","project":"openstack/shade","branch":"master","hashtags":[],"change_id":"I1b0d95fd1eae8f0a901c788b2d6f4884c37eeec6","subject":"Fix code specification with pep8","status":"ABANDONED","created":"2019-09-03 09:51:51.000000000","updated":"2020-01-16 02:21:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2b2bfc4cc67cdc800ccae8e1a1be7d5e39cd3b01","_number":679736,"virtual_id_number":679736,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-09-03 10:50:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 10:50:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec30902b7744b8a96bda4dcadaaa2f6ca6fde9df","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-09-03 09:51:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"802dfbeaed919971c3aa2f8ec2a83fe4a7965446","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-09-03 09:53:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"95fa5d551ba665b7d6d3cf8fa127e7619a91629a","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-09-03 09:54:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"89e5647c5539a850a78d84762187eacc37e3d222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 10:50:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca0336890b8b4ef59506ce5372956a54 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11453bdbcc494de78a943248a9b1bbbe : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5f00fc6879a4be088323c3f5ab587b4 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a940509e4df4d9a9d1d27eb6e34f0a2 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0183aa6a3614dd5a6a0e0957c3d3ab2 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8086d24faaf44996b5d3d3323dda4f58 : SUCCESS in 4m 26s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/cddecc4c931d497fae8da2522f941fbc : SUCCESS in 49m 27s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/5340aed0f4a048ed9b418ab0ac5b2b50 : SUCCESS in 51m 07s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/8fec0017d12c446eb255a0677d09e475 : SUCCESS in 4m 43s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/e80836e41dfb487aaf018a4adb3e16fd : SUCCESS in 4m 52s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/af46bcca16144bc7a2adb82323d1c279 : FAILURE in 5m 46s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/f52a6eb528674a288631dacff17f3377 : SUCCESS in 22m 44s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/dab1b2dafb1d4376a5e8556760427b81 : SUCCESS in 47m 30s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/412cd7c4ffb542a0b1c322fc42fbb01e : SUCCESS in 38m 01s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/6f41e0530f4244a4937d59f10c2d76e7 : SUCCESS in 49m 16s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d793ce649c8141f3b0cead781c8b4984 : SUCCESS in 49m 19s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/38ef24165e3a4e14bf19a202047bb20d : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":3},{"id":"2b2bfc4cc67cdc800ccae8e1a1be7d5e39cd3b01","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-01-16 02:21:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a7de691d0dac0dd95022bfdbe22d0d42f5370cc8","revisions":{"12f651e8bc6b60846fdfef552d7544d4e99e1db9":{"kind":"REWORK","_number":1,"created":"2019-09-03 09:51:51.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/36/679736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/36/679736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/36/679736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/36/679736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/36/679736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/36/679736/1"}}},"commit":{"parents":[{"commit":"ed25fe3d3617765f9a73e500aacce272011b815d","subject":"Switch openSUSE job name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/ed25fe3d3617765f9a73e500aacce272011b815d"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-03 09:51:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-03 09:51:32.000000000","tz":480},"subject":"Fix code specification with pep8","message":"Fix code specification with pep8\n\nChange-Id: I1b0d95fd1eae8f0a901c788b2d6f4884c37eeec6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/12f651e8bc6b60846fdfef552d7544d4e99e1db9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/12f651e8bc6b60846fdfef552d7544d4e99e1db9"}]},"branch":"refs/heads/master"},"f826ea7b885dc1a440709472bf391e793695139e":{"kind":"REWORK","_number":2,"created":"2019-09-03 09:53:50.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/36/679736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/36/679736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/36/679736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/36/679736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/36/679736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/36/679736/2"}}},"commit":{"parents":[{"commit":"ed25fe3d3617765f9a73e500aacce272011b815d","subject":"Switch openSUSE job name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/ed25fe3d3617765f9a73e500aacce272011b815d"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-03 09:51:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-03 09:53:36.000000000","tz":480},"subject":"Fix code specification with pep8","message":"Fix code specification with pep8\n\nChange-Id: I1b0d95fd1eae8f0a901c788b2d6f4884c37eeec6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/f826ea7b885dc1a440709472bf391e793695139e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/f826ea7b885dc1a440709472bf391e793695139e"}]},"branch":"refs/heads/master"},"a7de691d0dac0dd95022bfdbe22d0d42f5370cc8":{"kind":"REWORK","_number":3,"created":"2019-09-03 09:54:39.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/36/679736/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/36/679736/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/36/679736/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/36/679736/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/36/679736/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/36/679736/3"}}},"commit":{"parents":[{"commit":"ed25fe3d3617765f9a73e500aacce272011b815d","subject":"Switch openSUSE job name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/ed25fe3d3617765f9a73e500aacce272011b815d"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-03 09:51:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-03 09:54:26.000000000","tz":480},"subject":"Fix code specification with pep8","message":"Fix code specification with pep8\n\nChange-Id: I1b0d95fd1eae8f0a901c788b2d6f4884c37eeec6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/a7de691d0dac0dd95022bfdbe22d0d42f5370cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/a7de691d0dac0dd95022bfdbe22d0d42f5370cc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
