)]}'
{"id":"openstack%2Fshade~679738","triplet_id":"openstack%2Fshade~master~I4bf743fb9f239115adf98734dbc6674a052e3d44","project":"openstack/shade","branch":"master","hashtags":[],"change_id":"I4bf743fb9f239115adf98734dbc6674a052e3d44","subject":"Use requests replace request","status":"ABANDONED","created":"2019-09-03 09:58:41.000000000","updated":"2019-09-18 13:47:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"19be26ff77d863d0cccbaaa081bf0568addab7aa","_number":679738,"virtual_id_number":679738,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-09-03 10:59:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 10:59:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38233ec272adec2d0cbc92d65579385fd775cab1","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-09-03 09:58:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bf87dbc6b4d12600a748ca658e5b9d61c579a92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 10:59:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46fa182000ac48389312b29501eded98 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/410fcccdc9424df79cb0aba21e0f7a85 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7d3ea46805b4dc89069a28cfa9e1f55 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a707db400754830816b2cfae742fbf6 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34fcbd2ee67d45849f8da52c88e5f521 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d258cac5f9e44a6829e2f250beec058 : SUCCESS in 4m 13s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7a9eadc99c764169968709976baf4705 : FAILURE in 53m 01s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/f317abadf8cb4e74b03e9e823e272fbb : FAILURE in 47m 54s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/b091e180bae54aaf958f1b40beace0cb : SUCCESS in 4m 36s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/8193a685a55e4c548ba6024f6fe5ce16 : SUCCESS in 4m 27s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/9266ae0585a14dd7bb022b9448c2622c : FAILURE in 4m 38s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/fced39b7f09b466d853347c218775089 : SUCCESS in 21m 31s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/fb0f46db5823419ba04050b8e78c65fe : SUCCESS in 42m 17s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/98506e985b2241daa9e08c1353f5fad0 : SUCCESS in 36m 44s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/113c1896b67c493daec5ca923a436a8a : FAILURE in 54m 57s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/55d74456c15f4dbc9530e99723c903bc : FAILURE in 47m 41s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/f3612f5941bf4bb1b7c309df355768de : FAILURE in 43m 26s","accounts_in_message":[],"_revision_number":1},{"id":"19be26ff77d863d0cccbaaa081bf0568addab7aa","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-09-18 13:47:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c85c2fc99fb4b529966eccc2960a8a56c0cac8a","revisions":{"8c85c2fc99fb4b529966eccc2960a8a56c0cac8a":{"kind":"REWORK","_number":1,"created":"2019-09-03 09:58:41.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/38/679738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/38/679738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/38/679738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/38/679738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/38/679738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/38/679738/1"}}},"commit":{"parents":[{"commit":"a7de691d0dac0dd95022bfdbe22d0d42f5370cc8","subject":"Fix code specification with pep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/a7de691d0dac0dd95022bfdbe22d0d42f5370cc8"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-03 09:57:47.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-03 09:58:01.000000000","tz":480},"subject":"Use requests replace request","message":"Use requests replace request\n\nChange-Id: I4bf743fb9f239115adf98734dbc6674a052e3d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/8c85c2fc99fb4b529966eccc2960a8a56c0cac8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/8c85c2fc99fb4b529966eccc2960a8a56c0cac8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
