)]}'
{"id":"openstack%2Fshade~684929","triplet_id":"openstack%2Fshade~master~Id03251f85c78b02ddb0abe38232bc6dc530542c7","project":"openstack/shade","branch":"master","topic":"constraints","hashtags":[],"change_id":"Id03251f85c78b02ddb0abe38232bc6dc530542c7","subject":"Update the constraints url","status":"ABANDONED","created":"2019-09-26 05:54:48.000000000","updated":"2025-09-17 17:02:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"41ea8991b00a60ba1369e6c826299ad158a0cea4","_number":684929,"virtual_id_number":684929,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-22 13:50:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-12-22 13:50:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-12-22 11:50:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-12-22 11:50:39.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-02 13:23:05.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-01-02 14:16:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-22 11:50:39.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"8e837194533289b16f781f149baa4360d5e49d27","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-26 05:54:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4eacff3c2c9e5b6ebc34e58192eca748bf19556","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 06:57:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef4cec8ecf4c49ef91be44980c3b0533 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68581c330ac4b8f93ca87faecea10e8 : SUCCESS in 2m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee7f609bf807457f84fce28f20693894 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c67cf6f00af34f0da58fb08b5eeeec7b : SUCCESS in 2m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d7e0832c1694e69bc19f071af2876b4 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e8f149a650a4f438817e7ee18131beb : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/528bc08bf77948f291521ee643f70f21 : SUCCESS in 5m 20s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/96f7417065cd45b2a875f041cf88c548 : SUCCESS in 57m 20s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/cfef157ce28149b0a41f94001aa7407d : SUCCESS in 55m 40s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/8578ff24eee34ef5a9375045765902c6 : SUCCESS in 5m 32s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/6b065744759b4b1c98ff93d94aacedd6 : SUCCESS in 7m 58s\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/f09c7451079847aaaaafc2fe65485f12 : SUCCESS in 49m 32s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/8f31b9f932bf48e4a95ecadf1d0d50a3 : SUCCESS in 46m 06s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/a4c512e0e6e241c183c3772863545579 : SUCCESS in 56m 41s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d739a7e0f7e647828ee984ba82898e15 : SUCCESS in 49m 29s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/1f83b6ddc56f495cb31c21211085d00c : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f33f3a202747cdd37ec7492709979fa9a18640b2","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-01-02 13:23:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6455ed5e4d16473d935f3f5eddcca65ea0a32b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 13:23:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38dc423e88033683df0eac4e55d421894aabc84f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 14:16:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e0ae6014a7a4badaf0b278bd91bf1d9 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add489edcc614aef9771b9cdd91546e9 : FAILURE in 2m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f7598e659e64613870007a5f9a98365 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5c23037b46547a89a8dd0c88da811b5 : SUCCESS in 2m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/112b30254c5149d8b93046f3ef962a5a : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a72c41070d749938b7ad41a96432bcc : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3d593bf4d664d1d90b596e25b0ea5a6 : SUCCESS in 3m 36s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f5410b2452144ce2a329853e857bfa80 : SUCCESS in 43m 21s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/cf3c11a6813c4a828c53f69ba5508fdb : SUCCESS in 46m 24s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/33aaabbb562a481c9cbd8832a2b58171 : SUCCESS in 3m 29s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/c5cec5fdc5f9457bb0e52bb1c11259c0 : SUCCESS in 4m 01s\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/b0335874ea2443ac8da2f4c1bac82258 : SUCCESS in 35m 21s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/31aa9aba90524d87b507238b582836a9 : SUCCESS in 42m 23s\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/70ff56b7b5ac429cbcecd6cd71ebfce3 : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1fbec2dc37027347691d1df1b6a107299fc7649a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-22 11:50:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"08277d4061bdf690bfe50de46bf093fdcca7d9dd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-22 11:50:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f345a0b749f59716c30a9898eb12d2d411d16d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 13:50:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6115f8dea79f4a37ab9b74edc59d2687 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9255fafbde274d78a63f809d7209ce94 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/545a318943ed4a3695eef1009b76d720 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59b1f266615b4fdcb5d42aecb071e1e0 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53b714e7397949ac96df5ec7a9889569 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31ca8d1d61847908bd9fb7c46965948 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9b27bf84c884d2082de1b8bd80d87ca : SUCCESS in 3m 54s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/f4fc447e508e429bbb40cc50b3b22805 : SUCCESS in 39m 45s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/f0862926d8db46db8e4a350b9d882897 : SUCCESS in 4m 40s\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/c92077699f1f41d096f9f1c4038555c9 : FAILURE in 37m 31s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/5845dd274a274f9696f7a2e472ff781f : POST_FAILURE in 39m 14s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9c0146834fae4e9b9b35b515da1dca79 : RETRY_LIMIT in 37m 49s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/4a0b4e66218840bc8b11890471c34262 : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":2},{"id":"41ea8991b00a60ba1369e6c826299ad158a0cea4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-17 17:02:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4da93e54b0ccd7d812296cb4a7586f11ef8c3d42","revisions":{"f7751f66c70c907e29558670234be43bb18d3c79":{"kind":"REWORK","_number":1,"created":"2019-09-26 05:54:48.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/29/684929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/29/684929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/29/684929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/29/684929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/29/684929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/29/684929/1"}}},"commit":{"parents":[{"commit":"ed25fe3d3617765f9a73e500aacce272011b815d","subject":"Switch openSUSE job name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/ed25fe3d3617765f9a73e500aacce272011b815d"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 05:54:44.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 05:55:31.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Id03251f85c78b02ddb0abe38232bc6dc530542c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/f7751f66c70c907e29558670234be43bb18d3c79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/f7751f66c70c907e29558670234be43bb18d3c79"}]},"branch":"refs/heads/master"},"4da93e54b0ccd7d812296cb4a7586f11ef8c3d42":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-22 11:50:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/29/684929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/29/684929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/29/684929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/29/684929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/29/684929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/29/684929/2"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 05:54:44.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-12-22 11:50:17.000000000","tz":0},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Id03251f85c78b02ddb0abe38232bc6dc530542c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/4da93e54b0ccd7d812296cb4a7586f11ef8c3d42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/4da93e54b0ccd7d812296cb4a7586f11ef8c3d42"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
