)]}'
{"id":"openstack%2Fshade~691162","triplet_id":"openstack%2Fshade~master~If6b5c69a6c7f47b13afbd3871bfcf6818e96f5b3","project":"openstack/shade","branch":"master","hashtags":[],"change_id":"If6b5c69a6c7f47b13afbd3871bfcf6818e96f5b3","subject":"Switch to Ussuri jobs","status":"MERGED","created":"2019-10-25 02:59:10.000000000","updated":"2019-12-06 17:38:57.000000000","submitted":"2019-12-06 17:35:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691162-1575653757793-0f63802f","meta_rev_id":"41bd8c46be7dfba05ab1fc90445831acce27967c","_number":691162,"virtual_id_number":691162,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2019-12-06 15:12:41.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 17:35:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-12-05 15:21:14.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2019-12-05 15:21:14.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-05 15:21:14.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-12-06 15:12:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-06 17:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2e18612b68326f56620df208d577173dc832083","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-25 02:59:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b5f9a6542193acc1d0e4d6cce4bcf5bcb02c729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 03:52:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4f48e57f5af4cbfbcd4e2331744b190 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4469c3e9aaad465f818fb0776703fc11 : SUCCESS in 2m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ead4207357a84ad594ca2f1f79ef334b : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5307c8f11fe94ff98a63055c7a801ed3 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bad5abc0f9745d49f7fea3471835a99 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e2f55bca894936ab45726154e66886 : SUCCESS in 2m 57s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7355c9c490124dddac06ba30a761fea3 : SUCCESS in 47m 59s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/f840dc14eacd498681ab7a1f7641c444 : SUCCESS in 50m 15s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/022730c6778b4b67a07f21841be8f8b1 : SUCCESS in 4m 50s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/38385cc676ff4727ac694be1c6944e51 : SUCCESS in 4m 46s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/ca7b5396095f4bbaaba54e169421685e : FAILURE in 5m 59s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/9675cacdfae146fd89d29351a335ebad : FAILURE in 7m 20s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/990e306f32f84ee8836d3c2fae71b857 : SUCCESS in 37m 50s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/b428438b44ee40979d028e3a6f49aeac : SUCCESS in 35m 08s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/44ec53521360449c801e551d3a4d54da : SUCCESS in 42m 04s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9e0f544832d84fcb97dee18e57408fed : SUCCESS in 43m 23s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/bafedee4488f41049594614dcad8a229 : SUCCESS in 42m 06s","accounts_in_message":[],"_revision_number":1},{"id":"95da5094bae28e93cd7501750d5d98cc1bc9b350","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-12-05 15:21:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d5e1c574c3d6076083ed371ce0b40ea9bbf2495","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 15:21:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2fb7aac624133b148d0a9200c15289c6c7becce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 16:30:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dffa9025b57426f87733d650e9a7e99 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4ba4c1302d4fd9bb32a91d03256449 : RETRY_LIMIT in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86da15ca1f194100aee1ba912e0daacd : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc29104b676a4599ac4099fc89291e1e : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51ab5fe0d7d64f88b6effc7a8cdfa506 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763a1cb3ed1143a3a465dbdaeab5edbe : SUCCESS in 4m 51s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1145fad7af75429d851b5528c636091f : RETRY_LIMIT in 5m 32s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/66af88edde5745f78124faa496bd3f4c : SUCCESS in 48m 34s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/ea3a5b8d2e6445d5b648833526b01a8c : SUCCESS in 4m 52s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/4574f56b298941949804a76f7be8dd0c : SUCCESS in 5m 19s\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/382a50a93a5a43da93526cbda1b7c744 : SUCCESS in 43m 02s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/2caa7938cb5a40b4bdc2cf511af63d09 : SUCCESS in 44m 27s\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/999a6d273e074b0f8c90bc54de82d7d8 : SUCCESS in 48m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9dc3195c10720b5c306d15c7bbfa0df2865318b5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-06 15:12:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"901bc8964742b44784f413e1b8f4639c28d3858c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 16:25:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d5831aae5484246831a1e51b773ed95 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa9428f246e74506a3c6ff752dc1e0df : SUCCESS in 2m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5156e5a442f14b759ae89b2d8d32f60e : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82dc9b290f204d76ba5b8294ab92f191 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a05a4a569164a068e2205f0c40c9ca9 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18cc1a9d4f7c46568e261e097558cc88 : SUCCESS in 3m 55s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/31c84c4051ba44d7817eb865d7d0b2b9 : SUCCESS in 47m 23s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/4ccbd619306e4a258861f63623bbfd22 : SUCCESS in 47m 20s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/bce834b89e4448a0952eb16fae6adbbf : SUCCESS in 4m 17s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/b72e095d3fa3439d9761e514dfaa289b : SUCCESS in 3m 51s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/905cc276a1014314befbf5aa060d7dae : FAILURE in 8m 49s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/20cf95420ab64cedafb06a5b8dfcdec6 : SUCCESS in 22m 31s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/3b0c167b02a54f938b0deb2891035703 : SUCCESS in 59m 12s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/70398cb98bf34209b9357f172d6e08a8 : SUCCESS in 48m 19s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/7e75dd75b74e459591ddf288ee4b9733 : SUCCESS in 49m 50s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/78c4ed9af5b1406ba943863f76b17cf1 : SUCCESS in 51m 50s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/928689c7c8224e38b324b79fd4ae4641 : SUCCESS in 48m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c163004897bffd93b7ac878dfef8d7bfb3b00d25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 16:25:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28d6e067e2321c0a5f9046e2d0d79a7037d48487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:35:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/110506db12984251b70bcbb09bcfe262 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/545f8af166cf4b4aac7edda875a07909 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/606a29b27a024400a7d35a24ac817f73 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95c868060c4c4c72863f1a06686d76b2 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f17ff7bde43f474d9dbaff1595ca7605 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85af893848bd44a39011f8bee136af7f : SUCCESS in 7m 17s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d9e902b665554254b212c409731cb5a9 : SUCCESS in 57m 40s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/d8eb8899a1cc4aa888ed4b56ab781216 : SUCCESS in 1h 01m 14s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/51eda659c6914b048e87705fd71a1805 : SUCCESS in 8m 14s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/03e56cac62f343ee91c573f0b1a77408 : SUCCESS in 4m 43s\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/ef3beaa7e71342d786bb50509dce8658 : SUCCESS in 53m 15s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/b26c0e0bbc8641d9ae9aa3eb408e67f5 : SUCCESS in 55m 47s\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/9d74236263c74059a55ecebfc546788a : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0cb1cf3f777d7e8c4ad69657bfde19211eafa7e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:35:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"41bd8c46be7dfba05ab1fc90445831acce27967c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:38:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ba09bbe74142e3ac0824a2662c12e0 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e","revisions":{"e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e":{"kind":"REWORK","_number":1,"created":"2019-10-25 02:59:10.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/62/691162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/62/691162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/62/691162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/62/691162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/62/691162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/62/691162/1"}}},"commit":{"parents":[{"commit":"88b03d8ddd528935fa7e2a98198e7a858ea3243b","subject":"Add source and contributing link in readme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/88b03d8ddd528935fa7e2a98198e7a858ea3243b"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-25 02:58:56.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-25 02:58:56.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: If6b5c69a6c7f47b13afbd3871bfcf6818e96f5b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
